)]}'
{"id":"openstack%2Fqinling~661122","triplet_id":"openstack%2Fqinling~master~I02146bc67ec783786aa280f7759623abeb84bd21","project":"openstack/qinling","branch":"master","topic":"2005769/33486","hashtags":[],"change_id":"I02146bc67ec783786aa280f7759623abeb84bd21","subject":"Increase request timeout for sidecar","status":"MERGED","created":"2019-05-23 21:19:08.000000000","updated":"2019-06-17 10:29:58.000000000","submitted":"2019-06-17 10:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661122-1560767398761-9b2c7f29","meta_rev_id":"10a5a85acfc5b3c30bf3bf8591d0a638d3b61cd2","_number":661122,"virtual_id_number":661122,"owner":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-06-17 10:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-06-17 09:50:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-06-17 09:50:29.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 09:50:29.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-06-17 10:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3edf11c02f60cc6f9dbc531784c311ffc4a6f5ad","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-23 21:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"314c0e40d7bbde49be3434bcb1e13c8de3cf584b","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-05-23 21:20:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"148095d97555db0cecc2470f3bf1936f4f6264bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 21:59:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/22/661122/2/check/openstack-tox-cover/fbd566f/cover/ : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/661122/2/check/openstack-tox-lower-constraints/9ad7ffc/ : SUCCESS in 6m 23s\n- openstack-tox-pep8 http://logs.openstack.org/22/661122/2/check/openstack-tox-pep8/c7d9296/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/22/661122/2/check/openstack-tox-py27/f25dc24/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/22/661122/2/check/openstack-tox-py35/8aacd0a/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/22/661122/2/check/openstack-tox-py36/5eabce3/ : SUCCESS in 5m 25s\n- openstack-tox-py37 http://logs.openstack.org/22/661122/2/check/openstack-tox-py37/19a1449/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/22/661122/2/check/openstack-tox-docs/e731346/html/ : SUCCESS in 4m 26s\n- qinling-tempest http://logs.openstack.org/22/661122/2/check/qinling-tempest/ea1c900/ : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5d71d592b08e8de5496f9cda8e85b180f8e853a6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-06-17 09:50:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b8b60097bb79c8eaa34d5c875e9a4b086aedae6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 09:50:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac4313ab495143e9968a158328dc953944410947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 10:29:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/22/661122/2/gate/openstack-tox-lower-constraints/2a3d6da/ : SUCCESS in 5m 31s\n- openstack-tox-pep8 http://logs.openstack.org/22/661122/2/gate/openstack-tox-pep8/4083216/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/22/661122/2/gate/openstack-tox-py27/4d301a2/ : SUCCESS in 7m 13s\n- openstack-tox-py36 http://logs.openstack.org/22/661122/2/gate/openstack-tox-py36/5400490/ : SUCCESS in 7m 12s\n- openstack-tox-py37 http://logs.openstack.org/22/661122/2/gate/openstack-tox-py37/9b99eeb/ : SUCCESS in 6m 12s\n- openstack-tox-docs http://logs.openstack.org/22/661122/2/gate/openstack-tox-docs/89352f0/html/ : SUCCESS in 4m 16s\n- qinling-tempest http://logs.openstack.org/22/661122/2/gate/qinling-tempest/a7f1197/ : SUCCESS in 26m 23s","accounts_in_message":[],"_revision_number":2},{"id":"68093057e75783ce499a590e4e52dcbcf7989b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 10:29:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7234980037c368aa1673af78d72e1458e258d2c1","revisions":{"b5347407f0e0aaa43de43773432a62d55cf58590":{"kind":"REWORK","_number":1,"created":"2019-05-23 21:19:08.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/661122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qinling","ref":"refs/changes/22/661122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qinling refs/changes/22/661122/1"}}},"commit":{"parents":[{"commit":"de1d83bf59bcbb58d3c96759950619528739b19b","subject":"Use k8s 1.13 in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/de1d83bf59bcbb58d3c96759950619528739b19b"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-23 21:18:28.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-23 21:18:28.000000000","tz":-240},"subject":"Increase request timeout for sidecar","message":"Increase request timeout for sidecar\n\nBy default sidecar timeout to authenticate to Keystone\nis set to 5 seconds.\n\nIn most of the cases 5 seconds is enough but depending\nthe environment, the authentication complexity, etc...\nit could be too short.\n\nChange-Id: I02146bc67ec783786aa280f7759623abeb84bd21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/b5347407f0e0aaa43de43773432a62d55cf58590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/b5347407f0e0aaa43de43773432a62d55cf58590"}]},"branch":"refs/heads/master"},"7234980037c368aa1673af78d72e1458e258d2c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-23 21:20:52.000000000","uploader":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"ref":"refs/changes/22/661122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/qinling","ref":"refs/changes/22/661122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/qinling refs/changes/22/661122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/qinling refs/changes/22/661122/2"}}},"commit":{"parents":[{"commit":"de1d83bf59bcbb58d3c96759950619528739b19b","subject":"Use k8s 1.13 in devstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/de1d83bf59bcbb58d3c96759950619528739b19b"}]}],"author":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-23 21:18:28.000000000","tz":-240},"committer":{"name":"Gaetan Trellu","email":"gaetan.trellu@incloudus.com","date":"2019-05-23 21:20:34.000000000","tz":-240},"subject":"Increase request timeout for sidecar","message":"Increase request timeout for sidecar\n\nBy default sidecar timeout to authenticate to Keystone\nis set to 5 seconds.\n\nIn most of the cases 5 seconds is enough but depending\nthe environment, the authentication complexity, etc...\nit could be too short.\n\nChange-Id: I02146bc67ec783786aa280f7759623abeb84bd21\nStory: 2005769\nTask: 33486\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/7234980037c368aa1673af78d72e1458e258d2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/qinling/commit/7234980037c368aa1673af78d72e1458e258d2c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
