)]}'
{"id":"openstack%2Frally-openstack~1004226","triplet_id":"openstack%2Frally-openstack~master~I9af8e8e2dcb15503d2686fb88ab34643002cc2e5","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-09 00:26:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9af8e8e2dcb15503d2686fb88ab34643002cc2e5","subject":"tox: Drop redundant passenv","status":"MERGED","created":"2026-09-05 18:10:44.000000000","updated":"2026-09-09 02:22:09.000000000","submitted":"2026-09-09 02:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004226","meta_rev_id":"191e991cb1da0e888f074e03fc61d1e1da41ee5a","_number":1004226,"virtual_id_number":1004226,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 02:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-08 15:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-08 15:55:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 19:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 15:55:58.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"d7a9648a7188544689889ca1689f8ec926017677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f4ae1dca0a4f0051c5228e44cbda5500c7e7049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:11:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c91c63504b1756d2296fd108e3bea65542bbe45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:15:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"aa5d0616472f5a20092a678e53e68c8a9f99e21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 19:20:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2120c1ea7aa44afbbc65688e15a197f4\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a0dc74a1431048c0a13fc947a6cc322a : SUCCESS in 3m 29s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/1304b93213c94b8094b68b0fac7d61eb : SUCCESS in 2m 17s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/859a73be5126411eb476c1fe8a57adc0 : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a14ed7dbcbc64c689f9041847e43941c : SUCCESS in 2m 30s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/b444e602973f4b5ba6e675fadd3912ea : SUCCESS in 4m 54s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/45257ad121ca4e48b4dbab8c3d5b983f : SUCCESS in 4m 20s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/5a1eb167947c491c8b3e033543b23cde : SUCCESS in 6m 38s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/1f368e9c28d24374a94ce3b922217603 : SUCCESS in 17m 28s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/d951577b7352455dbd3f11333a7060a2 : SUCCESS in 6m 10s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/d23784123c9543e7b199e9f6a7120ec5 : SUCCESS in 16m 54s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/6546acfa7fe44a77a8f7c422485d9546 : SUCCESS in 54m 37s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/32faddcf79804a149a8ccd49d1530026 : SUCCESS in 23m 20s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/92924b63fc0f42a1af85df10835d1f1a : SUCCESS in 31m 24s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/474be68149f4420090fd373f04c9a80f : SUCCESS in 32m 38s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/3ad667278f3142febc941fc0cfb7b096 : SUCCESS in 36m 46s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/caf73a61ca5d437aa13abdbb3413952b : SUCCESS in 19m 22s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/b30e4e5511544372bb0f881c8ede5a27 : SUCCESS in 35m 55s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/d9152f2974174a64be13e4a12ad4b7df : SUCCESS in 56m 36s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/089b96b1ad554a26aefd07cc1d6705b1 : SUCCESS in 35m 15s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/43e694a50ff447eeb6b5408d91418e41 : SUCCESS in 36m 02s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/c625f07d651849b4ac8ff510f21d481d : SUCCESS in 17m 44s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/9f40602f47ae466f9638d18176b2ebe7 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":3},{"id":"a9e3f36dc5243b227ae7967535c2c16b8fd2216a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-08 15:55:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"84ef9d4097f79a36d320004ecb83649d5ccd7729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:56:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"81b27d891889f0052c2534498ee13866688fb6d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 17:07:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b03e6e8bd784e2a9c4251968eed2509\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5ae304099e1845549f802671918527e1 : SUCCESS in 6m 03s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/0d839d2e804741b7a50088e8451cbcce : SUCCESS in 3m 48s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebec83330c6b44b5a06a0b8935b63c4d : SUCCESS in 2m 29s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/531c148cf2784fc293c0a16851ea2913 : SUCCESS in 4m 09s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/566a8863a1174c6eb7386e27b6fdd80f : SUCCESS in 3m 51s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/1a21b128dec14eae9d05ed9676f7b276 : SUCCESS in 4m 09s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9cef1b4d74f14b6cb626f1c84b5008e9 : SUCCESS in 3m 46s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/a9c535b0122143eb8d875afa9cd15a33 : SUCCESS in 24m 02s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/67cadbfcf2fe417b926cf279aac5bd25 : SUCCESS in 8m 19s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/5d57172570854b73b6be5e85fa6a8e76 : SUCCESS in 29m 43s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/0d4346dbb0a7450b9a6826bc7a1a1292 : FAILURE in 36m 11s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/96fb1a1efe7144b586b6144cfc2f78c7 : SUCCESS in 18m 09s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/74236e606dfd4880af63bfc6164e015b : SUCCESS in 27m 11s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/ce4a5da788c14a16978ceef24cf04871 : SUCCESS in 33m 56s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/a31d57382b944411b65f3e35a405f0fa : SUCCESS in 34m 48s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/da632509e8c94204b9a26f4760c8a84b : SUCCESS in 1h 05m 37s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/f663aec3c1d64876916ebf646c38cbcb : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":3},{"id":"59f757b513cef7e005cd0ca541f27ff5326a1949","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-08 17:36:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d1ffe20528b5c12f8766367a771bcda2ee437439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:21:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9265e532307411d8785a1d7a711c020\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9a4405615590429d811aa0c8d89f9797 : SUCCESS in 3m 15s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/d13470aa7dd748a38b56e4c6ac259f79 : SUCCESS in 3m 13s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c44a8ed2cfc43688575cf1107de5b1f : SUCCESS in 1m 46s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/f95818ea91fb45b58580e00eb4e51913 : SUCCESS in 4m 07s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2e399e5bc9f9401aa25a0f8bd0c08842 : SUCCESS in 2m 30s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/3a326162aabe425dae7d2140ed59a7ee : SUCCESS in 4m 06s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/dacd3b75bc114020b4b16a57a0c2a062 : SUCCESS in 4m 01s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/3318506bea2d4546897ded773d2a5bec : SUCCESS in 17m 12s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/517a447caf79438fb4783f9e00b37d83 : SUCCESS in 8m 08s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/6f75dad02faf40f998784970c12c7d79 : SUCCESS in 29m 49s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/cd21c9a2754c4411bb842e399fa0ec27 : SUCCESS in 32m 21s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/2a2128c398f44598965b317dc95a285e : SUCCESS in 37m 43s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/c6b375695e024c25b7c804ad520b4c67 : SUCCESS in 38m 13s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/761b0aa733d54d2eaaf3498ead75745f : FAILURE in 16m 10s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/1ad2e8d172d3415e93df702e18f9dd54 : SUCCESS in 36m 04s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/b21de1c5ffa84be4bfd94f36c6aeb102 : SUCCESS in 33m 05s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/5d99c196242541cfb3b13ca9880c4713 : SUCCESS in 37m 48s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/705f831e80bb4a6693decfe84a15e178 : SUCCESS in 29m 17s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/ade5af9bb1df4b76bd7a7fb2e0feb511 : SUCCESS in 15m 07s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/2423a0a116d54c4e8373ace600eb793e : SUCCESS in 34m 52s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/7b03964454ee42cbac4643e481147614 : SUCCESS in 18m 25s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/5fb3df0cb43b442fbc08282386729ef3 : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":3},{"id":"e7e12caa8a59897a12764f9fc097fc6a62df94de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 18:22:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06cc4c7b4b2d3954d00ce9075aecad65b4cc6631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 19:19:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0e5ec127e434708a2f1ace5a5d84b16\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/338eb9b13a5a4c9095426853ef1753b9 : SUCCESS in 5m 50s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/32ac4ee9c4dd4feebaf8b7338f153548 : SUCCESS in 3m 46s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e422a31d80ee45d0a90940246b1f1540 : SUCCESS in 2m 11s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/0f897f091ad944429a57422d71a2bce6 : SUCCESS in 4m 24s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c42df281e8a845f687c45070c62f7b95 : SUCCESS in 4m 08s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/35ea005d20944bcab3dd97f2247cd9ca : SUCCESS in 4m 14s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/e6b8a72f91a843ff89bef22abb220723 : SUCCESS in 3m 40s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/ac55a8d90f1c47fd811a1d56930f6c32 : SUCCESS in 39m 01s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/e5517bde567e48df90376fab90d6012e : SUCCESS in 7m 59s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/e492a55540ab41d09082e40b13d441d6 : FAILURE in 11m 25s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/3d3550454dcc40bf92edb60a48cc567d : FAILURE in 11m 21s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/abb29519a94840ddb2aa4f320717e5dc : SUCCESS in 34m 31s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/fc923ad1430f427c94dc7117e478b084 : FAILURE in 22m 49s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/88f7fde41a664712ac8e89c388ff6681 : SUCCESS in 36m 06s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/0562b7e71d744690b86ff34273db18c7 : SUCCESS in 22m 49s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/9efacc12219a4ae68910b704d8207e73 : SUCCESS in 47m 02s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/ba45f7f36dd44760af8321a22968e882 : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0f0e43001948ad678a80d6d3ac24546b94667b2d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 00:26:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c09dfdb5a6faa8aea8dc4e93ebc9ccf193f219ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 01:30:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb54a4385a9e460084ecb4ed632574b5\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6c5aff68d8ba47cd84f47b6fbf1a9cde : SUCCESS in 6m 02s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/c753c5bbacdd4f938e6cff55afb5e315 : SUCCESS in 4m 11s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6210155d4e6c4458806a2bd4d9770a40 : SUCCESS in 2m 35s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/5f49c2ef49e340ed8e91fd0ca9c1d560 : SUCCESS in 4m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/01cc756b6112456e98e172fdb627965f : SUCCESS in 2m 18s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ec4382a5f2b949c7b457d4b1b37e25cf : SUCCESS in 4m 07s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/30638e784b3e4aefb8c1ce9425a823ab : SUCCESS in 3m 54s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/dd42d7759d70485e93906812fcd125ea : SUCCESS in 24m 16s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/7b840b97f0c348c19a0fcb3cf0591406 : SUCCESS in 8m 39s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/6f0ad4deb2f143ca9f2080f440b34487 : SUCCESS in 30m 48s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/9cb7375064434641a9f6fb326424bfb3 : SUCCESS in 51m 54s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/eb6ae211cc36482a821c912ea5204ff9 : SUCCESS in 32m 57s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/996ce352356842019c895a3b3a0ce255 : SUCCESS in 36m 18s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/121431af12f84d2aa5d38d25ec42858d : SUCCESS in 33m 59s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/7721abe9c0ac46f9a568d3803c0b4ecc : SUCCESS in 19m 09s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/b8ad0555207a4e40afea7659ab2657af : SUCCESS in 33m 08s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/8491c522628f4f21bfa478cb95bd27e0 : SUCCESS in 15m 36s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/92a96baba79a43c98b941ef17a5cbfe1 : SUCCESS in 57m 25s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/28c67a87cd2f494aad756acfed198ae9 : SUCCESS in 34m 35s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/ee92dc9d439e4cb4bd3c860d85b41921 : SUCCESS in 25m 13s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/bdf7c8f5e24c4dbb9fbdbc7622b90f59 : SUCCESS in 31m 01s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/9f05c4afd1c340e4b8da808a1fc78e61 : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":3},{"id":"559d13e55886ab9b0c8a909cd307148fcbfe5391","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 01:31:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a085b0ab866da7d192bee93e4cd6c172107d327f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 02:22:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa33cb5ffc447c5a99307eca6bce448\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b18b4f5d2b5d4019b8d5a3d358f028bd : SUCCESS in 5m 02s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/01982331c5eb41e5bc6e5c3c93f81575 : SUCCESS in 4m 39s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/90857ae06b5945349e2ad432738a7e1d : SUCCESS in 2m 32s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9c4cefd4f1134214aa81a624850cb40b : SUCCESS in 6m 16s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/1b13e8aec468471a9b14e397c343e868 : SUCCESS in 6m 11s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/a69167ecf1f74e8589fe0b7436393cc9 : SUCCESS in 5m 00s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/3f576533a0d245b1be5d1004883c606a : SUCCESS in 4m 10s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/510585d6e03143d3bec5ee917b48bec9 : SUCCESS in 27m 18s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/2d1105c707fa4c4783c5134bd74bf9d6 : SUCCESS in 10m 25s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/da6867f488f24abdaa4ce0533bd0a220 : SUCCESS in 20m 32s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/2fe0df31b251493b97c8910420289806 : SUCCESS in 41m 08s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/70608d03d7a04b13a084d5bec73dd73a : SUCCESS in 31m 23s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/64a20f23deb641ae97cd6dffc65c341f : SUCCESS in 24m 42s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/76c6cc84767843d6911a2fc030fb69ed : SUCCESS in 23m 21s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/7ae208e4bf03493a898d639bb23b7e40 : SUCCESS in 31m 45s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/ac309017451b4f9c9b9d08fcac547a20 : SUCCESS in 49m 05s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/ec9e8ee76da8427ba9d4a7b3145adfa1 : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":3},{"id":"191e991cb1da0e888f074e03fc61d1e1da41ee5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 02:22:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18e1843444bcef1fed9542b1c3e34b28b4d3dca1","revisions":{"14d476fc51374672a588799553c55bb4ffbee934":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:10:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/1004226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/26/1004226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/1"}}},"commit":{"parents":[{"commit":"779745bede90203b3be808fe2832bae17035fd8a","subject":"Merge \"Enable web-download import method for Glance Rally tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/779745bede90203b3be808fe2832bae17035fd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:10:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:10:32.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I9af8e8e2dcb15503d2686fb88ab34643002cc2e5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/14d476fc51374672a588799553c55bb4ffbee934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/14d476fc51374672a588799553c55bb4ffbee934"}]},"branch":"refs/heads/master"},"1d9fb663bdeaf8d75ac55740a203e30e719e8201":{"kind":"REWORK","_number":2,"created":"2026-09-05 18:11:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/1004226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/26/1004226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/2"}}},"commit":{"parents":[{"commit":"779745bede90203b3be808fe2832bae17035fd8a","subject":"Merge \"Enable web-download import method for Glance Rally tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/779745bede90203b3be808fe2832bae17035fd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:10:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:11:05.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I9af8e8e2dcb15503d2686fb88ab34643002cc2e5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/1d9fb663bdeaf8d75ac55740a203e30e719e8201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/1d9fb663bdeaf8d75ac55740a203e30e719e8201"}]},"branch":"refs/heads/master"},"18e1843444bcef1fed9542b1c3e34b28b4d3dca1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-05 18:15:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/1004226/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/26/1004226/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/26/1004226/3"}}},"commit":{"parents":[{"commit":"779745bede90203b3be808fe2832bae17035fd8a","subject":"Merge \"Enable web-download import method for Glance Rally tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/779745bede90203b3be808fe2832bae17035fd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:10:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:15:25.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I9af8e8e2dcb15503d2686fb88ab34643002cc2e5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/18e1843444bcef1fed9542b1c3e34b28b4d3dca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/18e1843444bcef1fed9542b1c3e34b28b4d3dca1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
