)]}'
{"id":"openstack%2Frally-openstack~1004731","triplet_id":"openstack%2Frally-openstack~master~Ie122b3cad8fc8ed83b0096035fc83170abd632b6","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 08:42:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie122b3cad8fc8ed83b0096035fc83170abd632b6","subject":"Drop warning filter related to pkg_resources","status":"MERGED","created":"2026-09-09 02:00:08.000000000","updated":"2026-09-10 08:42:42.000000000","submitted":"2026-09-10 08:42:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004731","meta_rev_id":"cd3fa1e30df3c8f02173d9a2a5347ddacac3efa6","_number":1004731,"virtual_id_number":1004731,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 08:42:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-10 07:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-10 07:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 03:22:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:42:49.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"1f16cb60af5747b90c6d34471d120b33626d87fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:00:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d485fb3e7864a15bc0d9360df8b61da601cab99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe5d6d922f448e79c08f387fa4f05b6\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7eb8d51c78b64c429e0256711f05c908 : SUCCESS in 6m 29s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f83ebe96ec4345b38a4d9af4f38a31be : SUCCESS in 5m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a8f86946a549878895f1a32402c039 : SUCCESS in 3m 06s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2e410802faf548c7ad15ad89f3e1a9cf : SUCCESS in 6m 37s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e7447d6cd0c64ae08a5b0d4c7d1bc7cf : SUCCESS in 2m 45s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/4f80a645c15c40ca918c2520bec5589d : SUCCESS in 2m 55s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/60b4481efca3465f8dddf6c88621ebb0 : SUCCESS in 3m 01s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/57a10b9f7efb4486ab47cc52f40ad4a1 : SUCCESS in 38m 06s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/e1eed1c1b39d49eeaa774f5f04c5ad74 : SUCCESS in 9m 17s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/9f543ca4fff047a7a69aeda63e67cf6f : SUCCESS in 34m 18s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/fee539c1035a452d858a1e4aee6e25e4 : SUCCESS in 56m 07s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/314c48d2f97247c08d67293924b0ed48 : SUCCESS in 29m 01s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/de3e7f683819478bb33f01050e51f289 : SUCCESS in 39m 25s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/76c063723f7c4c3da1acea6cab978eeb : SUCCESS in 31m 44s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/e6dcb44e979d407b85b85bcd8ed79204 : SUCCESS in 40m 04s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/6fad029311c6430a80387f8ac06c4733 : SUCCESS in 19m 37s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/e3f22e42d1be4cf19a7568282b91d1bf : SUCCESS in 37m 30s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/49cd6f278baf4fbf8984adcdcc513678 : SUCCESS in 36m 00s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/63a18e78bd5b4882823fa2b38bebde46 : SUCCESS in 20m 41s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/d9362eb1bf564ba19e6d0de258c84859 : SUCCESS in 22m 00s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/c2fb32c308224d7a84b368254c10dd5f : SUCCESS in 18m 08s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/ab7f55be549845dc9d298ef7a865348f : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":1},{"id":"66823e0928909e025eb7048a1cfadd84ccfd6184","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-10 07:42:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3328f9f9ea7fe5b4f9c070c0b97aaf84db2d21ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:43:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87be7f735806a93fb546732003ebde279ef842b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:42:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2fcd04cb9834eaf9f8aab3eae35971d\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8774969cc67f4f4f97c9229cc76bb5ce : SUCCESS in 5m 50s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/7c139417d2574e7eaa8ad23039a5c080 : SUCCESS in 3m 46s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c092c42bc1f244e6a803ebfb55ebe8c3 : SUCCESS in 3m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/cfe7769ee2fc48618642d34f35e61a6b : SUCCESS in 4m 20s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/d575574d90b543f297d5a7275e782ba2 : SUCCESS in 2m 25s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/85ca86aad37b47e59b5ec2597902ad3b : SUCCESS in 4m 13s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/65fed49927b440a4835ab8f062df17cb : SUCCESS in 2m 28s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/8d833bc3a9a849aaab651dd604ab4f20 : SUCCESS in 35m 40s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/f609c5eebd37427097e31392cecddc56 : SUCCESS in 8m 06s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/5edc5a19a04b4008acaf1c4061c8c3e5 : SUCCESS in 14m 26s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/826d0f9ce3604fe29b4825a6fcb47355 : SUCCESS in 33m 38s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/fed1294be4434eecbf938e6c722123e2 : SUCCESS in 30m 57s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/0471383fec3a4ab1908982251bbf4dcf : SUCCESS in 18m 34s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/2154ceda39704081be51666600ca777f : SUCCESS in 20m 09s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/44305c00ce194b999f8f14a0c5be1a63 : SUCCESS in 22m 19s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/3c55f4a16030475bbe2bd89f7773ecd8 : SUCCESS in 30m 28s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/16aec0ca4e39456db66970b54ac4fdaf : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":1},{"id":"cd3fa1e30df3c8f02173d9a2a5347ddacac3efa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:42:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2836d77410a834d47b7f8b78071e161747e3c98e","revisions":{"2836d77410a834d47b7f8b78071e161747e3c98e":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:00:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/1004731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/31/1004731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/31/1004731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/31/1004731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/31/1004731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/31/1004731/1"}}},"commit":{"parents":[{"commit":"def8543dd62fb5d4c7f18b2b457ffe315755c37f","subject":"Drop warning filter for legacy EngineFacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/def8543dd62fb5d4c7f18b2b457ffe315755c37f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 01:56:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 01:56:46.000000000","tz":540},"subject":"Drop warning filter related to pkg_resources","message":"Drop warning filter related to pkg_resources\n\nThe module is no longer available in recent setuptools and remaining\nusage cause hard failures.\n\nChange-Id: Ie122b3cad8fc8ed83b0096035fc83170abd632b6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/2836d77410a834d47b7f8b78071e161747e3c98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/2836d77410a834d47b7f8b78071e161747e3c98e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
