)]}'
{"id":"openstack%2Frally-openstack~1004758","triplet_id":"openstack%2Frally-openstack~master~Ie4ec245f62ab9db35bfa12810567434a851c8528","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 08:42:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4ec245f62ab9db35bfa12810567434a851c8528","subject":"Remove warning filter for imp module","status":"MERGED","created":"2026-09-09 03:08:01.000000000","updated":"2026-09-10 09:09:56.000000000","submitted":"2026-09-10 08:42:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004758","meta_rev_id":"b71d38ef2e69e39c966812adf79c4abedfe4af3b","_number":1004758,"virtual_id_number":1004758,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 08:42:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-10 07:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-10 07:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 05:35:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:44:07.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"a0daf3c3b827c12d5f3e4c90a4f4babd6f828ab1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 03:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"774b4a8f242650919eb23bc8998c1b64bf9ac51e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 05:35:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065ca3c6eb4a4d1e8c73fedb5d8ce1d8\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ed1029267e0345808c1a01a3c2603f5b : SUCCESS in 7m 37s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/750ee1968328419f8d0cacd8bf3549ad : SUCCESS in 5m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/38efe5090c804cb88ee6efb27881d65f : SUCCESS in 4m 14s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8394e2148a2d41e38b8ac40db924288b : SUCCESS in 4m 45s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/cdff473f23d84419acb15e53d90938e8 : SUCCESS in 5m 15s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/e2d80e61c8c74c1e8ead8b99bf3e42a9 : SUCCESS in 3m 29s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/72f8ffc3998f4bd69a0cd6e703ae3632 : SUCCESS in 3m 16s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/5b3ccc9ed5a64ea89569a6d95e4054f8 : SUCCESS in 38m 29s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/57f5828f7f5744ff84aa5ce8e194715f : SUCCESS in 7m 09s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/4306960de281464884f17f577fd3cbcc : SUCCESS in 19m 38s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/1ad3eb3ef4b94ab6a7c194815f8061a1 : SUCCESS in 52m 40s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/b8e3a8a158c14ab78e634b68bf486377 : SUCCESS in 49m 42s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/6d437ec2426244339f882bd3de17461d : SUCCESS in 38m 18s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/ca02f0ce88214c32baa2711ceb5b2a00 : SUCCESS in 19m 52s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/32e9d6225b0a44f4a4eb28414ba82e00 : SUCCESS in 38m 37s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/a12d474e72c3465eab796141e1d1daf4 : SUCCESS in 15m 04s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/1b7bd1100e374a04a790d780c3df53f8 : SUCCESS in 25m 21s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/e110065a0f8d49cd84ee00890669f04f : SUCCESS in 55m 42s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/dcf88da807ce49fbb0b9a9dab8468586 : TIMED_OUT in 2h 04m 50s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/3da91dd774da4fc3879712dc9ae88f0d : SUCCESS in 40m 42s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/9654d86d95354fffbbe8f205a294cf62 : SUCCESS in 34m 43s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/82312cd4f10a43f8b6bf72edd3073ccc : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a4df723c9a5daea01aab0880ff4685dc25f22ca2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-10 07:44:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc6402fb5512e59a2728c667dd88701f46026ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:44:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f165559a359435c8e9496277778c8246637631e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:42:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57f49addc4744747abed4e7b0a5a1f68\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/4eee8ba30f434476b2aa6b31b9cf7fc5 : SUCCESS in 5m 49s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/15bbc9b9a1104f7ead90b5df44003d35 : SUCCESS in 4m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e8f46d5411409f905c2d1d1952604c : SUCCESS in 2m 24s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/afaefbce6aa84f3295ca97aa48714b96 : SUCCESS in 7m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/5dcc40891f95404487a8355e9422f941 : SUCCESS in 4m 19s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/cfe19b319b9041749250219e213e2bd8 : SUCCESS in 4m 11s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/97242283e643436eb18744a5ae14d5ad : SUCCESS in 4m 08s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/23fc564ab6784ce09df92230faa91fa5 : SUCCESS in 33m 29s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/5d5b26117313485cb1406ba910253b9d : SUCCESS in 7m 40s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/427cae4a5ac3419487e31cbdcc625bd6 : SUCCESS in 14m 06s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/8e5258a2f23c4ffdb5c4b7f19e6ec5b0 : SUCCESS in 52m 19s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/7e0ba670e6374bf8b2d206b21ed8a892 : SUCCESS in 32m 17s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/2da0bdf51aeb404c832ff70468226ca0 : SUCCESS in 22m 50s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/8fe7ac16d39b473894c5488bc122c0a6 : SUCCESS in 25m 40s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/c8e0a69983ff45ceb3da9e8648bfc4e7 : SUCCESS in 26m 06s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/a06a770a7e234ffc92d63b0c607b888c : SUCCESS in 36m 51s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/48235f78832c49768b4f8ec4d1576aaa : SUCCESS in 40m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f10bdf86afb1746d72dda46a11846bc59053c5f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:42:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b71d38ef2e69e39c966812adf79c4abedfe4af3b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 09:09:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33990e758e9d107796afc098428ab846f06451fe","revisions":{"33990e758e9d107796afc098428ab846f06451fe":{"kind":"REWORK","_number":1,"created":"2026-09-09 03:08:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/1004758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/58/1004758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/58/1004758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/58/1004758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/58/1004758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/58/1004758/1"}}},"commit":{"parents":[{"commit":"8121848b1c806df97851fa424c5395d9b0cfb82d","subject":"Remove warning filter for oslo.db \u003c 14.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/8121848b1c806df97851fa424c5395d9b0cfb82d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 03:07:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 03:07:38.000000000","tz":540},"subject":"Remove warning filter for imp module","message":"Remove warning filter for imp module\n\nThis module was removed in Python 3.12.\n\nChange-Id: Ie4ec245f62ab9db35bfa12810567434a851c8528\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/33990e758e9d107796afc098428ab846f06451fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/33990e758e9d107796afc098428ab846f06451fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
