)]}'
{"id":"openstack%2Frally-openstack~1004915","triplet_id":"openstack%2Frally-openstack~master~I51c92c93ebd590e5779adda3d17cca6a607b1595","project":"openstack/rally-openstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-09 18:53:04.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I51c92c93ebd590e5779adda3d17cca6a607b1595","subject":"Use built-in StrEnum","status":"NEW","created":"2026-09-09 18:04:28.000000000","updated":"2026-09-10 04:01:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdb9dd893d155dbf9a69a0981e22e8eac2da08c6","_number":1004915,"virtual_id_number":1004915,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-10 04:01:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 18:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1b9a1ad4d25cc9b21a6ced9ab514543786efc63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 18:04:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11b44c3896a983d1c26f16bc9477e6696581ca4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:53:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f76b91677b3a481c9eeb01a3c9cd735b\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/eaa8fd360eff49368ed7988fba9fddea : SUCCESS in 7m 07s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/1c90215f02e4487c863bfca7b74820be : FAILURE in 5m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/10caa23180304ddea4d5e1e366930544 : SUCCESS in 2m 43s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/88a748cdd24247018b47620020f2ddf2 : FAILURE in 3m 53s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/1c8bed9d4dd0401ba7f9623dd91413fb : SUCCESS in 2m 52s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/0243f10da639445c90acff1fff40a616 : SUCCESS in 3m 46s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/cef00b0e1a324f12b5cd7183df6fb406 : SUCCESS in 3m 56s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/e4ee48fda89b42e79ee525750a67c482 : SUCCESS in 36m 06s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/bcdf0d9946ae4cf9b96d4660e9c4883d : SUCCESS in 7m 09s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/11b111d88b794b1d81e4f6777fa525a9 : SUCCESS in 27m 54s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/33c34329235d40198c7b40f0b7ee6ead : SUCCESS in 30m 29s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/0b491bca5d7243939f96ee5f3c61833c : SUCCESS in 45m 29s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/717c336259db48c39ad28c65d9be7b70 : SUCCESS in 38m 04s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/771019761899419ca1b0286b030f9318 : SUCCESS in 19m 27s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/5b580d857b7f4373a049c5a0a4daf98c : SUCCESS in 35m 32s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/744fda7b528548fdb86cc7bbf70cbfbb : SUCCESS in 34m 44s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/3fe9cb130620409c951025b282f25831 : SUCCESS in 34m 41s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/8fb7d1e8cad34019a51775c7a0744bdf : SUCCESS in 33m 20s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/98c505ba6fee4465933dd188075b8cdd : SUCCESS in 32m 23s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/39b7137b4f3c4563a8efe91559dfdc05 : SUCCESS in 18m 37s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/6412722ae3c44a07a04afa645292581c : SUCCESS in 17m 42s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/df7fafbe4a5b4931a8bbfaf74ccf6e59 : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5e4ca9c4c6f04c01a520f3dd9ae8b10ce923a59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:23:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d440b25d2966f230b79aff7f6adabc16edd971e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:24:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bdb9dd893d155dbf9a69a0981e22e8eac2da08c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:01:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5809304640944e6abd76c2e0f571cf02\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/50ba3f3faf4f489ab126a3d56ebc405d : SUCCESS in 6m 49s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/bffe0b8a8ce146469c237ba8d9185b5f : SUCCESS in 4m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d652e6c870c04ff48445d189c2df7a87 : SUCCESS in 2m 49s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/c4bd9a5ce3a743ba9cbae83732f05819 : SUCCESS in 4m 25s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/4a2083695ffc4f19b05c4361ce587297 : SUCCESS in 5m 04s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/05c781a5dfa24d7187b3747899741915 : SUCCESS in 4m 31s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/7dc256e26eb840fa9857f7a6e02ce740 : SUCCESS in 36m 41s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/41ecb06aa5944f41824949f845c3d7fa : SUCCESS in 8m 36s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/67321300f9e647a6ad01f3fc1ad1eb7c : SUCCESS in 31m 23s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/1767684be73d41009832b6d89900bb74 : SUCCESS in 50m 14s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/0e7e214fc4ab4145bc6243692b55800d : SUCCESS in 47m 37s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/dc3c4cdce0594623bc525eb1829e324b : SUCCESS in 37m 44s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/64f4297034ed4c56b9ba0e79b488f224 : SUCCESS in 19m 12s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/da20bb7ba482402b8f2518cd3e5b1855 : SUCCESS in 26m 58s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/6ef71f4e3126439ea6ef4224174277e3 : SUCCESS in 36m 01s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/fd8ac8b507fc4f3e8f65824a5e71d576 : SUCCESS in 26m 29s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/e41aa988fffa415da4fd905dc00f38ea : SUCCESS in 57m 45s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/e9187347779d47df8e8660edcb0b2167 : SUCCESS in 17m 47s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/0a28be98d7ce4fdcaab03baa2357ea79 : SUCCESS in 22m 53s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/d8482514db0b49deb4193152d21e59e6 : SUCCESS in 21m 21s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/91b6187c5ed94208aec9bb7089fae067 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c46f6b11dc57c3d7c17adbcbed39392e3d858f60","revisions":{"051df67f0bd1996928dec46642460a4e177d6b0b":{"kind":"REWORK","_number":1,"created":"2026-09-09 18:04:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/1004915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/15/1004915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/1"}}},"commit":{"parents":[{"commit":"779745bede90203b3be808fe2832bae17035fd8a","subject":"Merge \"Enable web-download import method for Glance Rally tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/779745bede90203b3be808fe2832bae17035fd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 18:03:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 18:04:18.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nPython 3.11 introduced this built-in type to replace the popular mixin\npattern.\n\nChange-Id: I51c92c93ebd590e5779adda3d17cca6a607b1595\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/051df67f0bd1996928dec46642460a4e177d6b0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/051df67f0bd1996928dec46642460a4e177d6b0b"}]},"branch":"refs/heads/master"},"66e99f2a07c6008de7d83c40f65c374b1ffe627e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-10 02:23:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/1004915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/15/1004915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/2"}}},"commit":{"parents":[{"commit":"dc0bb48e52610d12fdabe3594a24b73febfd5003","subject":"Remove Python 3.10 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/dc0bb48e52610d12fdabe3594a24b73febfd5003"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 18:03:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:23:26.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nPython 3.11 introduced this built-in type to replace the popular mixin\npattern.\n\nChange-Id: I51c92c93ebd590e5779adda3d17cca6a607b1595\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/66e99f2a07c6008de7d83c40f65c374b1ffe627e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/66e99f2a07c6008de7d83c40f65c374b1ffe627e"}]},"branch":"refs/heads/master"},"c46f6b11dc57c3d7c17adbcbed39392e3d858f60":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-10 02:24:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/15/1004915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/15/1004915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/15/1004915/3"}}},"commit":{"parents":[{"commit":"eebbd11d937832b19a4c74d9de0c56fc75d82b80","subject":"Remove Python 3.10 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/eebbd11d937832b19a4c74d9de0c56fc75d82b80"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 18:03:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:24:34.000000000","tz":540},"subject":"Use built-in StrEnum","message":"Use built-in StrEnum\n\nPython 3.11 introduced this built-in type to replace the popular mixin\npattern.\n\nChange-Id: I51c92c93ebd590e5779adda3d17cca6a607b1595\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/c46f6b11dc57c3d7c17adbcbed39392e3d858f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/c46f6b11dc57c3d7c17adbcbed39392e3d858f60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
