)]}'
{"id":"openstack%2Frally-openstack~1004962","triplet_id":"openstack%2Frally-openstack~master~If85bca902621db0cd108cd4f5571b04212bff0b7","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 11:47:44.000000000","reason":"Change was submitted"},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-09-14 10:48:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"If85bca902621db0cd108cd4f5571b04212bff0b7","subject":"Update job definition to reflect separate ceilometer plugin","status":"MERGED","created":"2026-09-10 05:07:50.000000000","updated":"2026-09-14 11:47:44.000000000","submitted":"2026-09-14 11:47:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004962","meta_rev_id":"55b5f25bbf7eff8d8246be055b10c9a353a5eea8","_number":1004962,"virtual_id_number":1004962,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-14 10:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-14 10:48:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 06:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:25:33.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"e4f891105b23ea0a99b5e3be6008772f64a2cf9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 05:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd837984cebe5096ab343abd63dd6d39238d9004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 06:07:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25dbb7c6257d4479a6b11cc68cade97c\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bf51245f9019447bbfddb5e32b1a51fd : SUCCESS in 6m 29s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/2b0faee57a644ae0af260e3a3ac0d2fa : SUCCESS in 3m 52s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c776f934934b5db5507f4c98fff87c : SUCCESS in 2m 33s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/a72c6366b6b14f3ca49541b24f9dfaf8 : SUCCESS in 4m 02s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/45e2063de1ed440c8ec5568a9d746c2f : SUCCESS in 4m 08s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/eddd7f0ed4914495826d55923dc86c23 : SUCCESS in 4m 00s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/2e97b7d3eb8a41bd85393a0eeecb8663 : SUCCESS in 37m 00s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/71e688fcf0ae4ba6b6d4064f3bc5821b : SUCCESS in 8m 12s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/2fc56c8a354b45fc8385364627a435de : SUCCESS in 28m 22s\n- rally-task-barbican https://zuul.opendev.org/t/openstack/build/e7566b8c39824bf88dd02b5f137114c4 : FAILURE in 34m 42s (non-voting)\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/3f15651ac05846c2bcce473ddca6be30 : SUCCESS in 50m 27s\n- rally-task-fwaas https://zuul.opendev.org/t/openstack/build/0372b98718044a598efe3bb0e4319e66 : SUCCESS in 30m 38s (non-voting)\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/22fb552f325f4ff39773643a59b91a94 : SUCCESS in 44m 14s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/70d2f96a378943fab7da1b6c59bb9583 : SUCCESS in 36m 34s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/8027d8e552874810baf1806eefd05f99 : SUCCESS in 22m 00s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/8ede6c8dc8824530b6b761c58128e9cc : SUCCESS in 37m 10s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/a8dbd47bb4fe4adaa376898f0f97ae0e : SUCCESS in 31m 52s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/4e1477cbea6d463291fb3e220c5a1c72 : SUCCESS in 20m 14s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/85453de1760643868a72901f6155c334 : SUCCESS in 42m 26s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/0877e2d3dfd74b5fa7974e6c4840a090 : SUCCESS in 33m 31s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/bab44ffec5ba431fb09aa5322006ff27 : SUCCESS in 26m 42s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/b5d5778373824e21b057dd66c2c7ac52 : SUCCESS in 31m 34s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/bf6ebef595554252a529bf04965ae3cb : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5b180170c51efefcdff858a461ff9f1288de4cd7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-10 07:25:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"381aa45a3512847a9abf23afe344c5976955d04c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 08:14:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6a3179bee3c62c148275f5fd6fac6acd6251a623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:14:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c17c784c792483380bebd861a69973b\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/71850871b3104b13a21fafdd30082cf8 : SUCCESS in 4m 28s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/881233edead149f182f92542e3c3cfcf : SUCCESS in 4m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42d0eccdafc4d78b6136b926b30f91c : SUCCESS in 2m 20s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/4299604383df48f4b94bd70fddd28ca6 : SUCCESS in 3m 55s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/24e7cf7b26f346f3b1a0c709b5d254be : SUCCESS in 4m 06s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/2d85117445344d0db358a6f0973642a7 : SUCCESS in 2m 27s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/fcbf113f973840fbb7e6c7faa173d36a : SUCCESS in 2m 18s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/6bfa5a0816994579996bf7a19c0dab66 : SUCCESS in 37m 47s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/bcf71ebebafd4fc1b4d26c5921cc2d32 : SUCCESS in 7m 34s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/cf22e4df06654d18a2d6850d0d8b772e : SUCCESS in 28m 33s\n- rally-task-barbican https://zuul.opendev.org/t/openstack/build/90c0f4a3881e496baa53d17702215114 : FAILURE in 30m 12s (non-voting)\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/27807b9388e943d3b07121eefe777dd6 : SUCCESS in 52m 23s\n- rally-task-fwaas https://zuul.opendev.org/t/openstack/build/fa7ef584d2c44a30bb99ad2959bb103d : SUCCESS in 20m 40s (non-voting)\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/0844fd73de004311bff0aaf50f4fb748 : SUCCESS in 48m 38s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/aabf565bc06d498bb8965d5e2518b870 : SUCCESS in 38m 45s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/08df6e1f5c0d49ef9b8ac88062e41190 : SUCCESS in 38m 59s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/a0b1b3e4911742fc9dc5758d5eeb85e7 : SUCCESS in 38m 03s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/5ff8d347ff3d47a3b174038f81b75e09 : SUCCESS in 21m 08s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/0884437c72e442c599b55ae31fabd862 : SUCCESS in 36m 50s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/b7c7ac43958d4feeae74738c247d65ee : SUCCESS in 35m 32s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/a9fc445787c5492bbc32c90edc66c326 : SUCCESS in 21m 14s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/e17e5ae970384ccb8e5ae681401db00e : SUCCESS in 21m 46s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/90970b99053c4d81908955aa505c8739 : SUCCESS in 18m 23s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/83f058cb4c384800a28226301b7e8e5c : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"687fbf8ff43fe8cbe973034d3d7ffd35b111c9a7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-14 10:48:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2920cf0fb321da7c24a6da2601c2e64de00a42f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:48:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9523a7e907a702a6e55b0b80775fd55da8e4b5e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:47:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f929532f5e7e4daab48406664b841da0\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/140011c7cc194ef2b9ea32b4c5b617b9 : SUCCESS in 6m 01s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/14b88704e76f40d4b2f75a44790dfe05 : SUCCESS in 3m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb12c4412fe5438a9f1546b3e4bcd1e6 : SUCCESS in 2m 30s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9ec461ec92094c8381a73457b76faeff : SUCCESS in 4m 04s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/f908f61ae7654cc3887b54c8329a86a1 : SUCCESS in 4m 48s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/aa0daf8a2ddc488294cd7b72865826cb : SUCCESS in 4m 21s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/70d82d6d8701428995de251ec12567d7 : SUCCESS in 4m 02s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/88b2c54ef1ad4adab10ca3e1ebcd355e : SUCCESS in 33m 37s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/70a56ab0165440ca85005230cd5a5ee9 : SUCCESS in 9m 15s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/a7a8acce3fee48659d73f935a0158d79 : SUCCESS in 26m 12s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/e7e1ae219f1a4f29ada630d6f4267640 : SUCCESS in 40m 03s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/14fd7cf38f6340579ca57d6d34080d46 : SUCCESS in 31m 05s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/62849c1f2c2b4f098de50f3f02faffeb : SUCCESS in 37m 34s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/86297fe8b7484aeeb5d0673ee336e3f2 : SUCCESS in 31m 16s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/6b3985ac4be4470ba6a7eab8a99d96fc : SUCCESS in 40m 36s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/1360a3dd770144f1a178950476462e27 : SUCCESS in 57m 38s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/48c044b417bc47a293fef4e7debcea96 : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":2},{"id":"55b5f25bbf7eff8d8246be055b10c9a353a5eea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:47:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d062c4c95a9e95bacd2ae253a8e495c63f4462f0","revisions":{"72116ef5ff917ebc30c1383609f4eff47d4b029b":{"kind":"REWORK","_number":1,"created":"2026-09-10 05:07:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/1004962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/62/1004962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/1"}}},"commit":{"parents":[{"commit":"c46f6b11dc57c3d7c17adbcbed39392e3d858f60","subject":"Use built-in StrEnum","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/c46f6b11dc57c3d7c17adbcbed39392e3d858f60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:04:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:07:32.000000000","tz":540},"subject":"Update job definition to reflect separate ceilometer plugin","message":"Update job definition to reflect separate ceilometer plugin\n\nInstallation of ceilometer was split out to its own devstack plugin.\n\nAlso drop redundant options to enable cinder because cinder is enabled\nby default.\n\nChange-Id: If85bca902621db0cd108cd4f5571b04212bff0b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/72116ef5ff917ebc30c1383609f4eff47d4b029b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/72116ef5ff917ebc30c1383609f4eff47d4b029b"}]},"branch":"refs/heads/master"},"d062c4c95a9e95bacd2ae253a8e495c63f4462f0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-10 08:14:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/1004962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/62/1004962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/62/1004962/2"}}},"commit":{"parents":[{"commit":"18e1843444bcef1fed9542b1c3e34b28b4d3dca1","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/18e1843444bcef1fed9542b1c3e34b28b4d3dca1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:04:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 08:14:19.000000000","tz":0},"subject":"Update job definition to reflect separate ceilometer plugin","message":"Update job definition to reflect separate ceilometer plugin\n\nInstallation of ceilometer was split out to its own devstack plugin.\n\nAlso drop redundant options to enable cinder because cinder is enabled\nby default.\n\nChange-Id: If85bca902621db0cd108cd4f5571b04212bff0b7\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/d062c4c95a9e95bacd2ae253a8e495c63f4462f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/d062c4c95a9e95bacd2ae253a8e495c63f4462f0"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c46f6b11dc57c3d7c17adbcbed39392e3d858f60","ours":"72116ef5ff917ebc30c1383609f4eff47d4b029b","theirs":"18e1843444bcef1fed9542b1c3e34b28b4d3dca1","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
