)]}'
{"id":"openstack%2Frally-openstack~1004963","triplet_id":"openstack%2Frally-openstack~master~I4e46c614e3b740533078bd88ba3d7e9952a378d3","project":"openstack/rally-openstack","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 06:07:31.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4e46c614e3b740533078bd88ba3d7e9952a378d3","subject":"wip: Enable heat tests","status":"NEW","created":"2026-09-10 05:07:50.000000000","updated":"2026-09-10 09:32:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0bb473dce55c0f402c8f42aecc8910840204f7b8","_number":1004963,"virtual_id_number":1004963,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-10 09:32:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 06:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d3ab3e3ddb9b47f07ffb2837928668e41e8d205d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 05:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5e78cd4db44672a99a3ba9cd10fb18151970bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 06:07:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/750719423b3b4b608c37eee0cc92d0da\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b5725279eeb04ebd81e93af0a2d5c954 : SUCCESS in 5m 56s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/c16625f6c4e84ef8aac07c1ac5dfdd01 : SUCCESS in 2m 45s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fc918e7a6884566bf829de4d2f08bef : SUCCESS in 2m 26s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/96b2d6ccdcce461c96b5dde452df4c12 : FAILURE in 3m 26s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/7f802f54f6b24dd999566f5f41a05f96 : FAILURE in 5m 08s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9864228e3b774a1780ecc7b915d182d5 : FAILURE in 4m 09s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/248238ce614b472187800e45da412b58 : SUCCESS in 38m 25s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/11f5699f608842fdb17da41df361d90a : SUCCESS in 7m 04s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/e7f91219626c485998cb5ec9c4a721a3 : SUCCESS in 12m 51s\n- rally-task-barbican https://zuul.opendev.org/t/openstack/build/2d9ec8e95c35439f91c9acaf81249f8a : FAILURE in 31m 51s (non-voting)\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/eb347507ee4c405eb45cc2df422ce7c8 : SUCCESS in 52m 06s\n- rally-task-designate https://zuul.opendev.org/t/openstack/build/48bbbaf83d8044b698048b240580d645 : SUCCESS in 19m 51s\n- rally-task-fwaas https://zuul.opendev.org/t/openstack/build/19b3ccb2bb5f4552bdba81c6e85516f2 : SUCCESS in 34m 21s (non-voting)\n- rally-task-heat https://zuul.opendev.org/t/openstack/build/34d402a5e76f44608b4d1bbd0a0d04b0 : FAILURE in 41m 34s (non-voting)\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/d61377aadda34652bd7188f4fa9a7e2b : SUCCESS in 34m 27s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/469c332fad1247fda4dde279c8dd52e0 : SUCCESS in 35m 04s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/bc4c4f7094374a4f96326c9943ed7b0f : SUCCESS in 20m 16s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/4dc9b4e3f5994924b15826db99534d67 : SUCCESS in 39m 45s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/c5c537805bf240e688abb8143fd1d9c2 : SUCCESS in 19m 34s\n- rally-task-neutron-trunk https://zuul.opendev.org/t/openstack/build/c579643a62574c7e9997334a4f533315 : SUCCESS in 23m 09s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/8727310b224849d9820ba94df8ff4490 : SUCCESS in 34m 34s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/b15957b37bee46d7b73362693252909f : SUCCESS in 43m 10s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/974192da8a654153869289cee77a630c : SUCCESS in 32m 14s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/2a1369ea54264e8c9d2ed25092359760 : SUCCESS in 21m 54s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/43e2a45742064fa7b8e04fb9d22c9d89 : SUCCESS in 20m 52s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/d8283f5e5d974b7592488725714a26a6 : SUCCESS in 39m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7c2949815285a118f0d6e630648754020a39e9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 08:14:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0bb473dce55c0f402c8f42aecc8910840204f7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 09:32:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7773293a68834d48934f55f1744938ae\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6213d43307b44e34aa5681f67aa8cb32 : SUCCESS in 5m 49s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a448310fa37b4a338c4bcc4e643e19fe : SUCCESS in 3m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fb030c619a47a799e140d86edafb80 : SUCCESS in 2m 12s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8590a7060caa4a8db69dec16284cff84 : FAILURE in 3m 59s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/6f21f5bd8e1643869c762d0a97b86c19 : FAILURE in 4m 02s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/965daea76a9848d282d55def8ffa79a0 : FAILURE in 3m 56s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/06b994eb52c64455a412bd436a272197 : FAILURE in 3m 39s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/033120f718da458584cf63f11cc072da : SUCCESS in 21m 41s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/92d5800c93de4811b1310703b970a60f : SUCCESS in 7m 16s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/a82c908fd6d04c8ba44a6977bceeac6a : SUCCESS in 19m 34s\n- rally-task-barbican https://zuul.opendev.org/t/openstack/build/ad4a419435ef41edaad3049977dccf23 : FAILURE in 19m 53s (non-voting)\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/24c715b128564fc5ac4ce51cfbc3fcc6 : SUCCESS in 51m 30s\n- rally-task-designate https://zuul.opendev.org/t/openstack/build/d0fe0ce33bc04775ae4beef89b482026 : SUCCESS in 28m 55s\n- rally-task-fwaas https://zuul.opendev.org/t/openstack/build/870dfca4ff424aabbd34267cc5224b3b : SUCCESS in 20m 00s (non-voting)\n- rally-task-heat https://zuul.opendev.org/t/openstack/build/425afb24ad13418b9ef5ac3386f2fa68 : FAILURE in 31m 18s (non-voting)\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/72c9ac3dc9b54fcb8052f53b99f6656a : SUCCESS in 50m 58s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/4bfd1346e0314865836d4f7a4e19d322 : SUCCESS in 23m 18s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/02c66e16b75446d4a81c613858c9a238 : SUCCESS in 36m 21s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/061d4409cd4042d68814d9b3c018459d : SUCCESS in 22m 33s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/c8dbabbbbc314d5995ec453eaffb1ae4 : SUCCESS in 20m 19s\n- rally-task-neutron-trunk https://zuul.opendev.org/t/openstack/build/4ab1703d8b23444db4e94c90fa0bcc0d : SUCCESS in 46m 05s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/e6f3d908b1c646e29c782a30e19aaea1 : SUCCESS in 21m 23s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/34bea2d43aad45e3b6afb6f1a5cae3b8 : SUCCESS in 1h 00m 17s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/06c0fe170c644f3c9cc05d13209c446a : SUCCESS in 30m 47s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/e6abeeac0e4a462a8b01b36e8d33d4d6 : SUCCESS in 35m 54s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/713f35455d2b4ff582f5dc5702820b7b : SUCCESS in 35m 04s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/06d1f66641264d4bafa29b7c250ec9a4 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f53ad0d6c9740b044f683696cae92788e439d9d","revisions":{"5b1824386cfcca5a0c428334587404a3301c778d":{"kind":"REWORK","_number":1,"created":"2026-09-10 05:07:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/1004963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/63/1004963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/1"}}},"commit":{"parents":[{"commit":"72116ef5ff917ebc30c1383609f4eff47d4b029b","subject":"Update job definition to reflect separate ceilometer plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/72116ef5ff917ebc30c1383609f4eff47d4b029b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:06:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:07:34.000000000","tz":540},"subject":"wip: Enable heat tests","message":"wip: Enable heat tests\n\nChange-Id: I4e46c614e3b740533078bd88ba3d7e9952a378d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5b1824386cfcca5a0c428334587404a3301c778d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5b1824386cfcca5a0c428334587404a3301c778d"}]},"branch":"refs/heads/master"},"9f53ad0d6c9740b044f683696cae92788e439d9d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-10 08:14:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/1004963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/63/1004963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/63/1004963/2"}}},"commit":{"parents":[{"commit":"d062c4c95a9e95bacd2ae253a8e495c63f4462f0","subject":"Update job definition to reflect separate ceilometer plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/d062c4c95a9e95bacd2ae253a8e495c63f4462f0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 05:06:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 08:14:26.000000000","tz":0},"subject":"wip: Enable heat tests","message":"wip: Enable heat tests\n\nChange-Id: I4e46c614e3b740533078bd88ba3d7e9952a378d3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/9f53ad0d6c9740b044f683696cae92788e439d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/9f53ad0d6c9740b044f683696cae92788e439d9d"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"72116ef5ff917ebc30c1383609f4eff47d4b029b","ours":"5b1824386cfcca5a0c428334587404a3301c778d","theirs":"d062c4c95a9e95bacd2ae253a8e495c63f4462f0","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
