)]}'
{"id":"openstack%2Frally-openstack~1005425","triplet_id":"openstack%2Frally-openstack~master~I44492bddb1532b76a73e52400e556bb068ade483","project":"openstack/rally-openstack","branch":"master","hashtags":[],"change_id":"I44492bddb1532b76a73e52400e556bb068ade483","subject":"Fix type hints for resource types","status":"MERGED","created":"2026-09-13 09:35:54.000000000","updated":"2026-09-13 11:23:53.000000000","submitted":"2026-09-13 11:23:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005425","meta_rev_id":"a024ea625d5f34cc19d41b43cd42a4886137233e","_number":1005425,"virtual_id_number":1005425,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-13 11:23:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-13 10:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-13 10:39:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 10:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc0df0f17fa97c8d3dffa53a57ba5e8788218f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-13 09:35:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4ee2de9e4e2dedef65c49d36b9984f66afa8e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 10:37:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc6c4e2802642688d3a3381d73bb42f\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f6da0630fcf64115969d7e79c908d51e : SUCCESS in 6m 22s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/58e751c69ab44355b84141a160d99c72 : SUCCESS in 4m 04s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2e767acf544819947d8cb0d3cb7856 : SUCCESS in 2m 59s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee5716dee53468e90cb723dfda54584 : SUCCESS in 3m 56s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e2958a213cd34a2fb4a6cc96e3369f9e : SUCCESS in 4m 01s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/898514cb52b04a12aea07e1d114dfd17 : SUCCESS in 3m 55s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/d532cdde330149beb5f8a05dad22846f : SUCCESS in 3m 44s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/bf26a3e146d641479d33d6840582110f : SUCCESS in 38m 08s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/112030a32a874c24a716479b6604e139 : SUCCESS in 7m 52s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/28644dc8149742fba60e43efed1852bd : SUCCESS in 18m 14s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/e920108e4a264ba8a34e330ee46c2c90 : SUCCESS in 50m 27s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/e112417c338b4145b003103b00fe7c95 : SUCCESS in 21m 59s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/30ab75858fae47a298989c8a171fe41d : SUCCESS in 35m 29s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/90b71f99f9b541b095ee1a88e1f8d4ae : SUCCESS in 15m 01s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/4ab66556fecd487995d586165863dfe9 : SUCCESS in 35m 05s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/5e08246beaf446d4847ff8856228a5f5 : SUCCESS in 21m 31s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/7667486601094d9796f4e31ad84d9416 : SUCCESS in 33m 24s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/b883a2ba05b54611a584cc66e4a00721 : SUCCESS in 59m 23s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/0dd5b3f9da8043fe9805be77040a915e : SUCCESS in 33m 44s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/30d81f9b1d69484b993e38718d1416a7 : SUCCESS in 33m 01s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/0b82bd9cda314f4f8381c2764aebe614 : SUCCESS in 29m 19s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/fdbd76ea20574df98882126e8658a60c : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":1},{"id":"2c3cb0702346225685ef00d1c9666f3597b07272","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-13 10:39:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b2659bdf1d3ec16dbb574f0b8ed44ac31e1ddf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 10:40:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a723c40c7d0259434129c7373be36893e6331ee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 11:23:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/186e1fefe48041ba8e5eb800c31896aa\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/934cf9231a614b7187a6b16b4324f6f8 : SUCCESS in 5m 57s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/dc9c2f98a4aa4fecb57ec839dcc8d332 : SUCCESS in 3m 20s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e800058bd90c424283c40c938113f9de : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/1a98236b2e0445ceb8e001731f61e97e : SUCCESS in 4m 58s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/7645c687257744b8b6001523ed056fad : SUCCESS in 3m 38s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/e431db6c161848bfb1f61d500b9b2a2e : SUCCESS in 3m 21s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/56f150555d1842f49a6f8b89ea1ff43d : SUCCESS in 2m 13s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/923fc5e5884b4435a63ee0d33d5b758d : SUCCESS in 35m 52s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/6bab1fb285fe42f7896c9c2b4c9c8887 : SUCCESS in 7m 57s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/73c6661cb59643c2b96e91914dd887e4 : SUCCESS in 29m 44s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/5550f81cba3a4c4f9afa313dd07b5f99 : SUCCESS in 36m 47s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/65c2177f4fc94b88b84c6137419f79e8 : SUCCESS in 32m 01s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/cd3fcfa516b642c6839f2b317ab0ec56 : SUCCESS in 24m 05s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/ebf4880d1e5d45f883e3aa6c6df3b7a2 : SUCCESS in 20m 36s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/9dc006de76f740ecb70029be9b25b21e : SUCCESS in 27m 53s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/30e0410863b646b5a92c3348030e75e5 : SUCCESS in 41m 11s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/a89fcdf984db4f578a1210286fa1342e : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a024ea625d5f34cc19d41b43cd42a4886137233e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 11:23:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55fc290ea659d859a25e79f1e59c33fdeefcbd35","revisions":{"55fc290ea659d859a25e79f1e59c33fdeefcbd35":{"kind":"REWORK","_number":1,"created":"2026-09-13 09:35:54.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/25/1005425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/25/1005425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/25/1005425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/25/1005425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/25/1005425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/25/1005425/1"}}},"commit":{"parents":[{"commit":"8a0360a229016c019791eb5ca1abe365c832c26c","subject":"Merge \"Remove warning filter for imp module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/8a0360a229016c019791eb5ca1abe365c832c26c"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-13 09:34:53.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-13 09:35:39.000000000","tz":120},"subject":"Fix type hints for resource types","message":"Fix type hints for resource types\n\nChange-Id: I44492bddb1532b76a73e52400e556bb068ade483\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55fc290ea659d859a25e79f1e59c33fdeefcbd35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55fc290ea659d859a25e79f1e59c33fdeefcbd35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
