)]}'
{"id":"openstack%2Frally-openstack~1005441","triplet_id":"openstack%2Frally-openstack~master~Ideb94c523f066a0a97e18f7c4edc4906d02245c2","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 11:49:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideb94c523f066a0a97e18f7c4edc4906d02245c2","subject":"Remove unused file","status":"MERGED","created":"2026-09-13 17:50:15.000000000","updated":"2026-09-14 11:49:29.000000000","submitted":"2026-09-14 11:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005441","meta_rev_id":"94d78157b60a502b7d1d1b2844195968b11a434d","_number":1005441,"virtual_id_number":1005441,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-14 10:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-14 10:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 18:51:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:48:19.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"bae2c5e60d608b382960af02468cc1b2c40b0351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 17:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63d77ab902045d737861b84af647b6ec73a3af48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 17:52:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"38358db15c89f1f5b7954aace89b377b9fca1589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 18:51:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5206cc8ad3044d389523a12b70f6be90\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a742ebee3fd4453ab8bf356c3db07d35 : SUCCESS in 6m 20s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/15efe60a41fa4b22880eabf1e1775815 : SUCCESS in 3m 34s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd74acf8e15043358b23ff52d946edd3 : SUCCESS in 2m 29s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/100f3973ef7045a0a084414d6d8f225a : SUCCESS in 2m 31s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/01c3aa715d1a4b82badfdfd3b727c5c7 : SUCCESS in 3m 51s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/aea32aca39d14cc4bf391b6974573034 : SUCCESS in 4m 00s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/ad1cb7b2c5a24adda4bea05b7ab69cc7 : SUCCESS in 5m 12s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/18d7bb209f67475cb0e61d19c8b88091 : SUCCESS in 35m 10s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/45b187fee4f8415887cec727d2e3e807 : SUCCESS in 7m 44s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/711494b8c6d444fabfa95d7d107bec27 : SUCCESS in 29m 30s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/d55d0132602840cba1afdb5eb08e60db : SUCCESS in 28m 10s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/69ba0db30fa240f18f8da8e4083c2bdb : SUCCESS in 46m 05s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/c601d139b8cb489c9d771af29fa76e30 : SUCCESS in 35m 40s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/354b60d1a36f4191b0d3e832070cc5db : SUCCESS in 32m 56s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/a2ec74289d724df7bdbed808cec58b03 : SUCCESS in 36m 45s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/d9cca678941142cbb1b30d256be66c48 : SUCCESS in 17m 23s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/33a4c53ea3284cfe8baffdfdd3247084 : SUCCESS in 22m 56s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/80d35f88f51042bca18f6a55f59b6d50 : SUCCESS in 55m 59s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/22e9ee926c6e49619c39c3367b5a1866 : SUCCESS in 35m 11s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/37c6bf0278534dc6bdb83cc8dc7bb7d0 : SUCCESS in 33m 20s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/1e3aaaa75114497bb2903b9f122a2d11 : SUCCESS in 19m 32s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/40d696253d3446a9afdd26cc04d374a3 : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":2},{"id":"52895641fc8212a7510a49c8538cdef818ffb665","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-14 10:48:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3579bce3a2a45f5b603376d826d1a0a48a66249f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:49:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14b21f7db80b0df01370e403042943e573d34008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:49:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f5dbb7b0b2a428d949668313a1ba497\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0b6c119cfd1a4e07be91a93e56386c06 : SUCCESS in 5m 57s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/5651403f04094f0a95ddb18823a81c33 : SUCCESS in 4m 22s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b2b122b554440bb116afe4082f0b91 : SUCCESS in 2m 32s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/214d4733e9cb461e9bd2447a71534adf : SUCCESS in 4m 07s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/aea8e71601d644029b3bc79b451a93e2 : SUCCESS in 4m 05s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/45cf063f90614f689bd45ebe2676be03 : SUCCESS in 4m 21s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/c92cfd32982743668588003e424ba512 : SUCCESS in 5m 22s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/f70bd81218f64b438232c87a41fa4cd7 : SUCCESS in 28m 42s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/25a6083506b044c8ac22a53719ff484d : SUCCESS in 8m 19s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/f542d5e0f7944cffaf1b54dfbf7dde25 : SUCCESS in 28m 11s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/86f4c4ac10154492b78481f85729b909 : SUCCESS in 51m 26s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/92d039704043442582963031e7e57925 : SUCCESS in 32m 28s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/a53864ee4c1c41299153faff1ced334e : SUCCESS in 34m 50s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/801857bb575b4d349c3e76513887cff7 : SUCCESS in 30m 45s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/98f2a5017180491eb13e307afab1082d : SUCCESS in 41m 51s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/b21ba02ec0bf4eaaa498ccb2dd038cd2 : SUCCESS in 55m 24s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/b49d6e54c0774038bcfcb2402f14450a : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":2},{"id":"94d78157b60a502b7d1d1b2844195968b11a434d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:49:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b04201077a43ebfb4b263acc3f4200c8b8ec6bb","revisions":{"fdfde78f4f489673d44524e6e9d178434715eaaf":{"kind":"REWORK","_number":1,"created":"2026-09-13 17:50:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/1005441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/41/1005441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/1"}}},"commit":{"parents":[{"commit":"779745bede90203b3be808fe2832bae17035fd8a","subject":"Merge \"Enable web-download import method for Glance Rally tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/779745bede90203b3be808fe2832bae17035fd8a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:49:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:50:07.000000000","tz":540},"subject":"Remove unused file","message":"Remove unused file\n\n/etc/motd has been unused since [1] was merged.\n\n[1] 9ca642e4e04d5aec48e7619439e25505ac32b213\n\nChange-Id: Ideb94c523f066a0a97e18f7c4edc4906d02245c2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/fdfde78f4f489673d44524e6e9d178434715eaaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/fdfde78f4f489673d44524e6e9d178434715eaaf"}]},"branch":"refs/heads/master"},"6b04201077a43ebfb4b263acc3f4200c8b8ec6bb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-13 17:52:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/1005441/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/41/1005441/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/41/1005441/2"}}},"commit":{"parents":[{"commit":"55fc290ea659d859a25e79f1e59c33fdeefcbd35","subject":"Fix type hints for resource types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55fc290ea659d859a25e79f1e59c33fdeefcbd35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:49:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:52:12.000000000","tz":540},"subject":"Remove unused file","message":"Remove unused file\n\n/etc/motd has been unused since [1] was merged.\n\n[1] 9ca642e4e04d5aec48e7619439e25505ac32b213\n\nChange-Id: Ideb94c523f066a0a97e18f7c4edc4906d02245c2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/6b04201077a43ebfb4b263acc3f4200c8b8ec6bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/6b04201077a43ebfb4b263acc3f4200c8b8ec6bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
