)]}'
{"id":"openstack%2Frally-openstack~1005442","triplet_id":"openstack%2Frally-openstack~master~I7bfb17848070c292a5c9eba7b8f1636a89593d8b","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-14 11:36:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bfb17848070c292a5c9eba7b8f1636a89593d8b","subject":"Drop warning filter for dnspython \u003c 2.0.0","status":"MERGED","created":"2026-09-13 18:02:34.000000000","updated":"2026-09-14 11:36:28.000000000","submitted":"2026-09-14 11:36:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005442","meta_rev_id":"cea4de0d8c59fcae6bd7e2b2ea1e52832fabca38","_number":1005442,"virtual_id_number":1005442,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 11:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-14 10:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-14 10:47:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 19:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:47:26.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"a0c104de43b4503847ebaedcef62c9028adf5969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 18:02:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52494ff8fd46225dfe3a18488bd7ee6ba3bbc59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 19:06:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7147cd6eec487b8bf324a5f8ba1091\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1ca87f75095d4696a165b3bd723eb626 : SUCCESS in 5m 55s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/7f3ecffeca6f40fca88994ed2345e0f0 : SUCCESS in 3m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa54bb890744c3797cafb9fd33210fa : SUCCESS in 1m 42s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/258a019a21424b90934183cb56fdb2f8 : SUCCESS in 4m 53s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/d4e78d2ee3904393ab1baabc23f4c518 : SUCCESS in 4m 25s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/655787ae04c64b21afaf766cded08b38 : SUCCESS in 4m 51s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/740fa0d7d0234ccdb7520da8c4f17a0c : SUCCESS in 3m 52s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/525b583d367a4d6199de33eecdca5e36 : SUCCESS in 36m 58s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/c052909fa24a419d91e81762c4c39808 : SUCCESS in 7m 31s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/a89c9af9091041458b80b92643640980 : SUCCESS in 20m 07s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/280dfd35ac0941899f1df78d884b4a38 : SUCCESS in 48m 11s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/4beb2114248e444fa71f0917e6487f87 : SUCCESS in 24m 35s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/166a9b1bce764f61b62cbaca456b6015 : SUCCESS in 34m 28s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/d3b2572876cd4756b8174780485f0c4a : SUCCESS in 34m 53s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/6079a58b37134ba589408197ae06993e : SUCCESS in 24m 11s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/3d2ce9818be0498f816fc79d2fec10dc : SUCCESS in 32m 14s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/976659e054de4f3a90c138b31488a304 : SUCCESS in 37m 53s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/e776c77f670a4ccfbee2ca0a093ba963 : SUCCESS in 57m 52s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/855296e4918544398ba89d62498a3161 : SUCCESS in 33m 26s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/4e0bb2a161bc47cd852a736f7af6445a : SUCCESS in 31m 22s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/d6ba980974604bdbaa1ebeefebe9443c : SUCCESS in 16m 00s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/f094aaeb02fc47afbd210fbbfa5f9d37 : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7722ec1fdfc65f34823d356370504db375472bfb","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-14 10:47:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ee50985226f0f2d0fd1bbcdc299571ce1e095f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 10:48:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ca84de97f0af781b9e1f1c26653d8c1b8dbea9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:36:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b27e486041a64dfe994912c650a0cc1a\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d3e5c5c0df49492f8f4260348a0c1d70 : SUCCESS in 5m 59s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/82d91f6bd87d45a7968b642b6d7b677f : SUCCESS in 3m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/659547a7dfa14d9a82b9ef9c9692b20d : SUCCESS in 2m 22s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/06bd06baea31473894164d08f4d5e1c4 : SUCCESS in 4m 37s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/67381d4fdaea49198c294db566c63932 : SUCCESS in 3m 22s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/cdfb1a3c424c4eb28e1513854f976bf9 : SUCCESS in 3m 38s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/125102e60dd648189164d33391393141 : SUCCESS in 3m 57s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/8dda4436b04049b393ad145258cfbc94 : SUCCESS in 36m 26s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/bc71a60698fe4ed7952700fd7645dbce : SUCCESS in 7m 50s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/9a3859fc5ccb4bdba7748fb5efaff7a2 : SUCCESS in 26m 30s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/c84615c70ae244ca80dd081aa410b664 : SUCCESS in 39m 12s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/3fe59910bc21451eacec0387aff241e9 : SUCCESS in 33m 04s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/5be306ba4f6f47379685f0bf467d0df7 : SUCCESS in 27m 22s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/039931911bee467289b18349caa0b16b : SUCCESS in 33m 23s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/acd2852bcdbc453eb05c90574edaa03c : SUCCESS in 40m 57s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/1e54dd4db9c74366bfe13943e2dde8b1 : SUCCESS in 47m 15s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/7ee9450eebc6455bb2992801cdc0cf2d : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"cea4de0d8c59fcae6bd7e2b2ea1e52832fabca38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 11:36:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f919e6c14c5253413f015475cee1da1f25663cb0","revisions":{"f919e6c14c5253413f015475cee1da1f25663cb0":{"kind":"REWORK","_number":1,"created":"2026-09-13 18:02:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/1005442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/42/1005442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/42/1005442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/42/1005442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/42/1005442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/42/1005442/1"}}},"commit":{"parents":[{"commit":"55fc290ea659d859a25e79f1e59c33fdeefcbd35","subject":"Fix type hints for resource types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55fc290ea659d859a25e79f1e59c33fdeefcbd35"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:59:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 18:02:26.000000000","tz":540},"subject":"Drop warning filter for dnspython \u003c 2.0.0","message":"Drop warning filter for dnspython \u003c 2.0.0\n\nThe dns.hash module was removed in dnspython 2.0.0, which was released\non Jul 17, 2020.\n\nChange-Id: I7bfb17848070c292a5c9eba7b8f1636a89593d8b\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/f919e6c14c5253413f015475cee1da1f25663cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/f919e6c14c5253413f015475cee1da1f25663cb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
