)]}'
{"id":"openstack%2Frally-openstack~1005778","triplet_id":"openstack%2Frally-openstack~master~Ie520875fabd27e89f6a0e3091268ea2405883b9c","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 20:46:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie520875fabd27e89f6a0e3091268ea2405883b9c","subject":"Remove leoftovers of monasca support","status":"MERGED","created":"2026-09-15 17:57:03.000000000","updated":"2026-09-15 20:46:13.000000000","submitted":"2026-09-15 20:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005778","meta_rev_id":"3068ff77240a1968ef12de461c4c0521383f58ad","_number":1005778,"virtual_id_number":1005778,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 20:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-15 19:51:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-15 19:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 18:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 19:51:37.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"2faf8f0e66a8f08fea8b100f6bff10b28d4799a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 17:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21821a8004e75a1414174bc163b10e0d14408b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:48:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6a976750644a45a83323b2d8e86254\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d2f6c52a87647bfa5da3cab71602585 : SUCCESS in 3m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9ce798a440cd4aa4b9415e0f25df3d8d : SUCCESS in 6m 06s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a7066ef06ae54bf1b520ec3f9e28c95f : SUCCESS in 3m 56s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/afedd58fc2744baf9727ee05a49cd937 : SUCCESS in 1m 57s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/9efb0804b1a94275976249d895503f76 : SUCCESS in 4m 05s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ef1e9d20aa454567815089fe62f15648 : SUCCESS in 3m 12s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/e54c7e21fffb4245afb36022e8680857 : SUCCESS in 5m 21s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/d02034e8a6224a6fbe644b93f62f1770 : SUCCESS in 6m 06s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/b8b49bebd177471a834963bd2bcd9954 : SUCCESS in 36m 50s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/e45218d73fa0405b9d027874cff7a696 : SUCCESS in 7m 46s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/fcca9f7781074efe9f71fc425515db84 : SUCCESS in 19m 15s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/511606fc9c9045e8ba82d9972c18006f : SUCCESS in 38m 57s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/ebf18187a4664de0bf2aba063c816d62 : SUCCESS in 27m 35s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/eb7e77fb1a084d3e847946853f23f643 : SUCCESS in 17m 30s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/39ae4b121149402c85b7756301461734 : SUCCESS in 23m 00s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/98ded2b9ab0443acabbfc9d46919e289 : SUCCESS in 18m 24s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/b087878e5ab348649d586e93a91503a3 : SUCCESS in 23m 27s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/1d4b1e7bb47a436794b48699b8a08d65 : SUCCESS in 34m 57s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/85539a93cc744154a9c4aab7dc770ba5 : SUCCESS in 42m 19s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/0a0e93a79ee54d42963b55c74f4b3d28 : SUCCESS in 31m 43s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/9349b5adddfa4034a27c20b802724165 : SUCCESS in 34m 42s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/8189eb682cc94e03bb6eec901ff99050 : SUCCESS in 28m 24s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/7c82b7fc54e0472b97ba8b4e3657eb74 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1a136b2ff35e820132e7a753a15637a745031a50","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-15 19:51:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5161aae9641348c6ba6f5b8d58f9c4f7269c780a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 19:52:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2097b49bcf66248c66ca366021b90a9cd1ef8d69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:46:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5dfa6816f6d406b9ad9cfd38777fb22\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cdc5af297bce4db88ab018b35da1adb7 : SUCCESS in 2m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2b63ef76822d4f45813e72b8bc9809af : SUCCESS in 3m 55s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f73602ef858646d2a4ebd3a7109ce87a : SUCCESS in 2m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4553f6adbb0a43b1bc6c3fce9a898511 : SUCCESS in 2m 33s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b40979e9206444df947282d65a4b8acc : SUCCESS in 4m 20s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/29a06b800fb14126a70a367152c2d145 : SUCCESS in 5m 22s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/92bfe7d50dcd4c8c8cc38c6f32b872e2 : SUCCESS in 5m 22s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9ea04931cece4ae6884feb7b21ca6b08 : SUCCESS in 4m 10s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/77a8cf0469e549a2950696dbc18e5d2a : SUCCESS in 36m 39s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/bcbccc2115f142d9ad9c2d5ef178bcf2 : SUCCESS in 7m 47s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/707e847321784fe79c89014c130caab1 : SUCCESS in 23m 25s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/225375e37fec44bdbf6db02c6ab6834a : SUCCESS in 52m 02s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/55ceb56843d64a289fdd8046925d8a76 : SUCCESS in 28m 27s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/b397580221b8437cbcd2617cd5b8eb44 : SUCCESS in 37m 12s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/ecaeca79dc5a4be9989f7cff992f491c : SUCCESS in 23m 44s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/b725cbe47e004fc993bdf2df45fcd236 : SUCCESS in 22m 24s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/925468150f4c4a77b6d436e24a710198 : SUCCESS in 44m 58s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/419e2d8c02384a3fbca9fc077f1e25d0 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3068ff77240a1968ef12de461c4c0521383f58ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 20:46:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbfce6a8981d023fc33c8f9b1e76a88f039f275c","revisions":{"bbfce6a8981d023fc33c8f9b1e76a88f039f275c":{"kind":"REWORK","_number":1,"created":"2026-09-15 17:57:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1005778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/78/1005778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/78/1005778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/78/1005778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/78/1005778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/78/1005778/1"}}},"commit":{"parents":[{"commit":"77e671dc790e5a9c910087799bbf603bbf58bf9a","subject":"Merge \"Rework glance layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/77e671dc790e5a9c910087799bbf603bbf58bf9a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:56:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 17:56:51.000000000","tz":540},"subject":"Remove leoftovers of monasca support","message":"Remove leoftovers of monasca support\n\nMonasca support was remove a while ago[1].\n\n[1] 284d08d562308b3f39507469a85c792029913aaa\n\nChange-Id: Ie520875fabd27e89f6a0e3091268ea2405883b9c\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/bbfce6a8981d023fc33c8f9b1e76a88f039f275c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/bbfce6a8981d023fc33c8f9b1e76a88f039f275c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
