)]}'
{"id":"openstack%2Frally-openstack~1005819","triplet_id":"openstack%2Frally-openstack~master~I821052c0b8d39b7c790bf25c148c7f896b7ada79","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-23 16:57:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I821052c0b8d39b7c790bf25c148c7f896b7ada79","subject":"Deprecate pre_newton_neutron","status":"MERGED","created":"2026-09-16 02:49:37.000000000","updated":"2026-09-23 16:57:51.000000000","submitted":"2026-09-23 16:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005819","meta_rev_id":"547ab0dc55cc6a5f8cd096ba49543a3369ce48ee","_number":1005819,"virtual_id_number":1005819,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 16:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-23 15:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-23 15:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 04:01:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 15:58:59.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"c6f24e66fca518b992a6d95ad157108b1d15d1bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 02:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e6b916c25d6eeacd6accdc4be9d687c471aeb3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 04:01:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c45a3b50ff4f67a454ff18fc00c216\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f9eb806164234953934b5f5beef7eb57 : SUCCESS in 8m 04s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/b023185c668c496bb44fd8c75b78f717 : SUCCESS in 5m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/44bc385e8b924a33adb171d252af0f8c : SUCCESS in 3m 56s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b4e336ba0bd849879e78769b57deb015 : SUCCESS in 4m 44s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/75a08b755e8745f3b90f1d4366b5bd12 : SUCCESS in 4m 32s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/b1a41a7dbbd641c3b2180a31a539fa7a : SUCCESS in 2m 51s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/83d6472e197844b8ae71f1da70bb2214 : SUCCESS in 4m 46s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/0c8e4c228fa24a85ae94ea63583f5e74 : SUCCESS in 27m 45s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/73fa16b0933e4299af12afb66f481019 : SUCCESS in 8m 09s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/ce364e16731545b99a414c519e88eaac : SUCCESS in 28m 14s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/d3fb182cbd514a3d9fbee3fc406529d5 : SUCCESS in 51m 45s\n- rally-task-glance-upgrade https://zuul.opendev.org/t/openstack/build/dd8999e9e0904ef8a3cc720c9e5800ec : SUCCESS in 41m 12s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/3e157f32772c4204ba57b92f5614e75b : SUCCESS in 26m 05s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/e5917004534245e79de8dc6aa7a266d4 : SUCCESS in 19m 10s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/1e92ba33a7b943b486fca8cda00acb8e : SUCCESS in 41m 53s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/2404bc7abceb4d348c760e4eda305c6e : SUCCESS in 24m 45s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/d22307779cf440e4aca8a4f6f9f21adc : SUCCESS in 33m 49s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/2151c746975f448f826ddb599266b4b5 : SUCCESS in 34m 33s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/26cf5a10d19340a8bfdc312e74698671 : SUCCESS in 32m 32s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/b726da6ec00d452e8fa9b45f25c3a608 : SUCCESS in 41m 05s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/e31453dc20da4ed2b7b54f40f4500e05 : SUCCESS in 37m 18s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/05fc90d092424de49ad9d7a3936e25e2 : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"77e81babc831b3bb53a119bb9917eb9d76532f09","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-23 15:58:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5db6b4120d5c14341ea1b60cae8fe5dc55e4f07a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 15:59:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81da9219d8d2bba9c1ab12e74e0f2a46dad5182d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:57:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab7ab3fd98e4d1e8a3906f4dd1cdf30\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e71f9476b6114e5cb3042856aa3d2876 : SUCCESS in 3m 52s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/8984dd836ce14520ac141174f8771eb1 : SUCCESS in 2m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23410c964cb4e20b624372ecfc3bcab : SUCCESS in 3m 53s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a5cf835032dd4b92a7827d2bcac33508 : SUCCESS in 4m 20s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/32b5dba106c1468ca004acadadbc336b : SUCCESS in 2m 33s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/6732a6b47c2348a68e91035d76aec6b6 : SUCCESS in 6m 11s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/15fa50d959e64474b1f27ff985bb28de : SUCCESS in 2m 20s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/5d6f3bffd19944729209fc03d130bd9e : SUCCESS in 35m 44s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/236a80ef260646da8050a7809ebc90e1 : SUCCESS in 9m 00s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/009fbe00ad364c32b9c34cb04e6e74d2 : SUCCESS in 25m 10s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/8975f3df80a74b7aa9acdc9ba4e6b7f8 : SUCCESS in 53m 04s\n- rally-task-glance https://zuul.opendev.org/t/openstack/build/028a72546d1c4af7922b39244e4fd324 : SUCCESS in 29m 09s\n- rally-task-keystone-swift https://zuul.opendev.org/t/openstack/build/294c73b3f66a43979bf5fc1e55ad2f28 : SUCCESS in 24m 43s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/1a59595e13064f2f8f2845a440a6031d : SUCCESS in 26m 24s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/ddccadc0516a4a2cad8828beaf198341 : SUCCESS in 48m 13s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/10f4e7ce798e4f81b7d2841ecefab4a0 : SUCCESS in 39m 50s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/94b5bd49cb014b0c828a77bc0a209b1c : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"547ab0dc55cc6a5f8cd096ba49543a3369ce48ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 16:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f93281f6a7b1254d06df05af9be9676aac428bd","revisions":{"5f93281f6a7b1254d06df05af9be9676aac428bd":{"kind":"REWORK","_number":1,"created":"2026-09-16 02:49:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/1005819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/19/1005819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/19/1005819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/19/1005819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/19/1005819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/19/1005819/1"}}},"commit":{"parents":[{"commit":"77e671dc790e5a9c910087799bbf603bbf58bf9a","subject":"Merge \"Rework glance layer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/77e671dc790e5a9c910087799bbf603bbf58bf9a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-16 02:48:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-16 02:49:25.000000000","tz":540},"subject":"Deprecate pre_newton_neutron","message":"Deprecate pre_newton_neutron\n\nNewton was released quite long ago and we don\u0027t expect any users may\nbe running it nowadays.\n\nChange-Id: I821052c0b8d39b7c790bf25c148c7f896b7ada79\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5f93281f6a7b1254d06df05af9be9676aac428bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5f93281f6a7b1254d06df05af9be9676aac428bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
