)]}'
{"id":"openstack%2Frally-openstack~1006002","triplet_id":"openstack%2Frally-openstack~master~Id115dc02478b4c93ad4ad915020f4e7944e49249","project":"openstack/rally-openstack","branch":"master","topic":"crash-on-sleep","hashtags":[],"change_id":"Id115dc02478b4c93ad4ad915020f4e7944e49249","subject":"[tests] Stop relying on global mock for time sleep","status":"MERGED","created":"2026-09-17 08:51:41.000000000","updated":"2026-09-17 10:31:30.000000000","submitted":"2026-09-17 10:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006002-crash-on-sleep","meta_rev_id":"18dbe6d9e3e836b8ec33668b3e1c6df68e6257e6","_number":1006002,"virtual_id_number":1006002,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-17 10:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-17 09:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-17 09:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 09:38:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2e9fe53d5da9d5c4a2f9b221ae374d40c472a3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-17 08:51:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0a3e412b8acc39c648b089c8ebc0d4c72439b10","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-17 09:12:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f190d12675f4fae58f59bd0a5ca0a1f797828ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:38:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a60c0e817d484443bfd46004164b84ee\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a0eeb16cccd04b42bf836fc9feaf2efe : SUCCESS in 7m 25s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/783178df61bb49449e455fa718b359bb : SUCCESS in 4m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c21a14eda4e4c06a65cf7910f73a006 : SUCCESS in 2m 27s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/cfff9eccf3f7468991e0e8b7ed14c6ac : SUCCESS in 4m 27s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/b9ab9dc0f9f44614bf572c4542991077 : SUCCESS in 5m 08s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/9fbf5dc23d8942358a2644314711a9c6 : SUCCESS in 4m 42s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/a6d9df544d1a4822b21f00e5588110ce : SUCCESS in 4m 29s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/e03fc83347ed4f188c1ef778a84f026e : SUCCESS in 8m 11s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/4c4cf13646694e018065f716e575ac7a : SUCCESS in 31m 04s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/cf563570b38842d9819809e184ac45f5 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7c5736f9c3a2b19e43ae9495c0f84a7f116e1514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 09:47:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8d3147dfd54569b552a515a6fc4b8f86fdbf34b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:31:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/647977bdc5b746f3a506f27b2253437b\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b044915b3db643198e016f978f03ee4d : SUCCESS in 6m 39s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/cc53b378994d4c5397ba0710566a7dce : SUCCESS in 3m 59s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8be4f2fe3b14d9cb161caf6696ad9c8 : SUCCESS in 2m 33s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/5d9d32013d1b4fb2846fa6a9fbf3dfad : SUCCESS in 4m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e3e86b7895b347c598a356a69b3f7a9a : SUCCESS in 4m 18s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ba984a49d30048038b252ef487b15956 : SUCCESS in 4m 27s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/a5db6e7ee8ab40979689d2f2e72cce41 : SUCCESS in 5m 17s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/fa977c388845475bb8757c2adbc976ee : SUCCESS in 37m 48s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/884ca9b4c9634b6aa9fc203a3ec1da3e : SUCCESS in 9m 22s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/3c0632a89a144a1b86bfe7fcad13f056 : SUCCESS in 29m 48s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/a87d696b56d845968be7add7dba8a516 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":1},{"id":"18dbe6d9e3e836b8ec33668b3e1c6df68e6257e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 10:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7330c7ccd11385261812f4dc7ce20ac901c6cd27","revisions":{"7330c7ccd11385261812f4dc7ce20ac901c6cd27":{"kind":"REWORK","_number":1,"created":"2026-09-17 08:51:41.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/02/1006002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/02/1006002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/02/1006002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/02/1006002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/02/1006002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/02/1006002/1"}}},"commit":{"parents":[{"commit":"06676fbe3159078731879709f5c1c87d271e2fdb","subject":"Merge \"Extend irrelevant files for functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/06676fbe3159078731879709f5c1c87d271e2fdb"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-17 08:51:11.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-17 08:51:11.000000000","tz":120},"subject":"[tests] Stop relying on global mock for time sleep","message":"[tests] Stop relying on global mock for time sleep\n\nChange-Id: Id115dc02478b4c93ad4ad915020f4e7944e49249\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/7330c7ccd11385261812f4dc7ce20ac901c6cd27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/7330c7ccd11385261812f4dc7ce20ac901c6cd27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
