)]}'
{"id":"openstack%2Frally-openstack~657306","triplet_id":"openstack%2Frally-openstack~master~I8f92ef87722d057514d0906d068ddaadd88c8758","project":"openstack/rally-openstack","branch":"master","topic":"fix_issues","hashtags":[],"change_id":"I8f92ef87722d057514d0906d068ddaadd88c8758","subject":"Fix various issues","status":"MERGED","created":"2019-05-06 03:03:25.000000000","updated":"2019-05-16 21:24:54.000000000","submitted":"2019-05-16 21:24:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657306-1558041894700-deb1f29f","meta_rev_id":"05e72f5f7176615fd5b4287ed86a2ea56dae06ca","_number":657306,"virtual_id_number":657306,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-05-16 21:24:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":2,"date":"2019-05-16 19:50:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":1,"date":"2019-05-16 19:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 01:23:50.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-05-16 19:50:04.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-05-16 21:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f393784f10d00e76c63a4d2d4824dccfe066fc7","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-06 03:03:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"300674803f0c33ea3c7e95fd9c3199ad9ce41d04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 04:37:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/fa074d0/ : SUCCESS in 3m 37s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/b83f2a9/ : SUCCESS in 4m 07s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/1ad73b7/ : SUCCESS in 5m 24s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/842790c/ : SUCCESS in 3m 36s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/a17fdab/ : SUCCESS in 4m 52s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/e5244ed/ : SUCCESS in 4m 11s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/951c05b/ : SUCCESS in 36m 19s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/e621905/ : SUCCESS in 15m 11s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/8033634/ : SUCCESS in 29m 11s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/792304c/ : SUCCESS in 36m 36s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/7b7f679/ : SUCCESS in 1h 05m 03s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/706cbdd/ : SUCCESS in 47m 40s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/4c8df68/ : SUCCESS in 45m 25s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/eaf62e6/ : SUCCESS in 39m 20s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/484a2f4/ : SUCCESS in 53m 53s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/b25c225/ : SUCCESS in 1h 23m 05s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/b31b078/ : SUCCESS in 38m 15s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/91f4567/ : FAILURE in 34m 41s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/3f3c3c7/ : SUCCESS in 1h 27m 04s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/c1333a9/ : SUCCESS in 43m 08s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/02759a5/ : SUCCESS in 1h 11m 08s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/26af96b/ : SUCCESS in 39m 59s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/7929bbc/ : SUCCESS in 34m 56s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/d0081d8/ : SUCCESS in 35m 49s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/051f14a/ : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6e6b0d5620ba1459b3e7540adffaebe272e9e808","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-06 11:11:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0fac5ed90a6c5ac9107b393c8a5618be1a0484a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 12:53:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/3e72ebb/ : SUCCESS in 4m 16s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/905ecbe/ : SUCCESS in 4m 13s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/4df013e/ : SUCCESS in 6m 07s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/5a826f9/ : SUCCESS in 4m 40s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/6cb3079/ : SUCCESS in 7m 44s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/67a7733/ : SUCCESS in 4m 53s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/f19c596/ : SUCCESS in 45m 18s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/aedb88a/ : SUCCESS in 4m 45s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/650ddad/ : SUCCESS in 33m 01s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/4dbbe8a/ : SUCCESS in 35m 02s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/e8b86b9/ : SUCCESS in 1h 09m 00s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/211ade5/ : SUCCESS in 57m 00s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/624ea3e/ : SUCCESS in 50m 10s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/5f716e2/ : SUCCESS in 37m 10s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/6f01db8/ : SUCCESS in 42m 13s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/a4b85e6/ : SUCCESS in 1h 36m 21s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/cbeaf96/ : SUCCESS in 35m 38s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/cebbd66/ : FAILURE in 42m 23s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/38505e1/ : SUCCESS in 1h 34m 29s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/80703a2/ : SUCCESS in 40m 32s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/581bfa3/ : SUCCESS in 1h 21m 21s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/f37a9d7/ : SUCCESS in 43m 18s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/a2fe492/ : SUCCESS in 36m 01s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/1df62ab/ : SUCCESS in 35m 04s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/cac7714/ : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5c6714eae1a378152332966c02cf634dfa7bd120","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-06 15:28:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5bced52c0be440de9608dce2325bf9bb32a30a8a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-06 15:28:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ffe179eb18866307df5cf7416888f603bb20be63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 17:25:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/0f3526b/ : SUCCESS in 5m 56s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/1354b4b/ : SUCCESS in 5m 04s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/60828a7/ : SUCCESS in 6m 35s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/2618d5c/ : SUCCESS in 4m 28s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/9bf8c37/ : SUCCESS in 5m 16s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/d22348c/ : SUCCESS in 5m 09s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/d2a7af5/ : SUCCESS in 42m 29s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/fcb0cd2/ : SUCCESS in 5m 16s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/65d99e3/ : SUCCESS in 35m 19s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/8620aea/ : SUCCESS in 36m 17s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/897b74c/ : SUCCESS in 1h 10m 43s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/df57049/ : SUCCESS in 1h 00m 03s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/8d7081a/ : SUCCESS in 46m 56s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/9e3a7aa/ : SUCCESS in 37m 52s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/7e9810a/ : SUCCESS in 42m 46s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/96c3bf6/ : SUCCESS in 1h 44m 37s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/beece53/ : SUCCESS in 40m 30s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/4045a20/ : FAILURE in 50m 48s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/b8e1568/ : SUCCESS in 1h 27m 42s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/6114200/ : SUCCESS in 43m 12s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/b299a2a/ : SUCCESS in 1h 12m 53s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/d8d5fa9/ : SUCCESS in 40m 14s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/29dd3bd/ : SUCCESS in 32m 15s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/3b98a39/ : SUCCESS in 34m 34s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/fecc2ec/ : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"15dd3fa1e86df6fb8e66db5c098b5479e46e7d0b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-06 20:23:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4f200126438d76e098e129cd27ed7ba4cc01c15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 21:53:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/d48d371/ : SUCCESS in 4m 19s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/062a65b/ : SUCCESS in 4m 06s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/2e534ef/ : SUCCESS in 5m 49s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/83b952a/ : SUCCESS in 4m 15s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/348fc1a/ : SUCCESS in 5m 28s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/d1d9499/ : SUCCESS in 4m 12s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/9ad0cee/ : SUCCESS in 38m 51s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/753f9c5/ : SUCCESS in 5m 07s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/54f1ce5/ : SUCCESS in 28m 43s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/58dc28b/ : SUCCESS in 35m 59s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/51b8473/ : SUCCESS in 1h 04m 20s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/0073243/ : FAILURE in 44m 48s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/01c42da/ : SUCCESS in 42m 01s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/c5f0835/ : FAILURE in 44m 38s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/ef96e6c/ : SUCCESS in 37m 47s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/37eb327/ : SUCCESS in 1h 26m 56s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/355c824/ : SUCCESS in 39m 16s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/eb9fba7/ : SUCCESS in 40m 05s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/48f1dce/ : SUCCESS in 1h 15m 50s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/93397bc/ : SUCCESS in 42m 34s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/502b30e/ : SUCCESS in 1h 12m 00s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/60bf3ea/ : SUCCESS in 41m 23s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/6f94613/ : SUCCESS in 37m 03s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/372cf32/ : SUCCESS in 35m 38s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/680e11b/ : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ee6f7a59b53c8f3e0b2015c2987ae3d0ffa5fd5d","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-07 00:23:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0769d37942f1b5a5fa12d753643cb93bcc8ebcaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 02:07:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/9e6af64/ : SUCCESS in 5m 41s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/159a34a/ : SUCCESS in 4m 59s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/918fbd0/ : SUCCESS in 6m 14s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/123b4cf/ : SUCCESS in 4m 05s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/70d134c/ : SUCCESS in 4m 56s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/3ae89d4/ : SUCCESS in 4m 33s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/cd1ba2f/ : SUCCESS in 43m 09s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/c7b0229/ : SUCCESS in 5m 55s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/4a83e59/ : SUCCESS in 33m 29s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/84433a2/ : SUCCESS in 35m 58s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/e2f8801/ : SUCCESS in 1h 11m 09s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/5c97c06/ : SUCCESS in 54m 52s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/e4197ed/ : SUCCESS in 41m 48s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/267971a/ : SUCCESS in 34m 42s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/089f60e/ : SUCCESS in 45m 59s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/7f3a72e/ : SUCCESS in 1h 38m 59s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/53caacd/ : SUCCESS in 41m 18s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/ba48dff/ : FAILURE in 49m 53s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/d3d611d/ : SUCCESS in 1h 23m 31s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/01c0634/ : SUCCESS in 43m 22s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/543016a/ : SUCCESS in 1h 13m 18s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/299db84/ : SUCCESS in 44m 13s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/53d4583/ : FAILURE in 46m 51s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/0628383/ : SUCCESS in 36m 13s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/9a62043/ : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8b8468b1416a2605e58d85a8384e9bb9975068dc","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-08 00:44:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"17f273390c5fed97da8f455640d6af8e3acd03d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 02:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/f8809a8/ : SUCCESS in 5m 01s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/49d64c7/ : SUCCESS in 4m 14s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/f39fc83/ : SUCCESS in 6m 59s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/41d8d0c/ : SUCCESS in 4m 08s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/642fe63/ : SUCCESS in 5m 54s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/5a99ab6/ : SUCCESS in 6m 11s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/04cf39a/ : SUCCESS in 38m 33s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/4e743f9/ : SUCCESS in 5m 53s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/eaabe5f/ : SUCCESS in 29m 59s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/01a4597/ : SUCCESS in 35m 11s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/f8537a6/ : SUCCESS in 1h 08m 04s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/2c35d67/ : SUCCESS in 1h 03m 27s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/afc80f0/ : SUCCESS in 57m 45s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/8768df7/ : SUCCESS in 35m 12s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/4010e25/ : SUCCESS in 51m 07s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/cad222c/ : SUCCESS in 1h 33m 29s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/cc773fa/ : SUCCESS in 33m 58s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/6e49f0b/ : SUCCESS in 45m 32s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/ae0e51e/ : SUCCESS in 1h 25m 05s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/73ae635/ : SUCCESS in 38m 55s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/fc2636f/ : SUCCESS in 1h 10m 44s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/0664560/ : SUCCESS in 33m 49s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/0c3cba7/ : SUCCESS in 31m 44s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/920a3a5/ : SUCCESS in 32m 54s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/d227d0f/ : SUCCESS in 44m 52s","accounts_in_message":[],"_revision_number":1},{"id":"845304c29f348cc15d155594498d0d6e5093573f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 02:23:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6967e9bff459462900694b2e5fc98afec1a53137","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 03:23:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/gate/rally-tox-pep8/e1d3d45/ : SUCCESS in 5m 10s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/gate/rally-tox-py27/6729ff2/ : SUCCESS in 5m 37s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/gate/rally-tox-py35/97aef3a/ : SUCCESS in 6m 04s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/gate/rally-tox-py36/26287bf/ : SUCCESS in 5m 39s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/gate/rally-tox-py37/63eae1c/ : SUCCESS in 5m 44s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/gate/rally-tox-cover/3fd8822/ : SUCCESS in 5m 41s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/gate/rally-dsvm-tox-functional/1ac095d/ : SUCCESS in 45m 02s\n- rally-docker-check http://logs.openstack.org/06/657306/1/gate/rally-docker-check/fc53dcd/ : SUCCESS in 5m 26s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/gate/rally-task-simple-job/a5200de/ : SUCCESS in 35m 02s\n- rally-task-cinder http://logs.openstack.org/06/657306/1/gate/rally-task-cinder/e32b6da/ : SUCCESS in 55m 36s\n- rally-task-ironic http://logs.openstack.org/06/657306/1/gate/rally-task-ironic/bf99a23/ : SUCCESS in 37m 12s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/gate/rally-task-keystone-glance-swift/923b746/ : SUCCESS in 50m 00s\n- rally-task-mistral http://logs.openstack.org/06/657306/1/gate/rally-task-mistral/704dae3/ : SUCCESS in 35m 17s\n- rally-task-murano http://logs.openstack.org/06/657306/1/gate/rally-task-murano/004aaf7/ : SUCCESS in 35m 02s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/gate/rally-task-neutron/2421bdb/ : FAILURE in 45m 38s\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/gate/rally-task-telemetry/3cbb665/ : SUCCESS in 37m 23s\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/gate/rally-task-zaqar/e4c92ab/ : SUCCESS in 31m 31s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/gate/rally-verify-tempest/3c14ee6/ : SUCCESS in 34m 05s","accounts_in_message":[],"_revision_number":1},{"id":"31ff12d0efdadb123f3493c9c0b00a8080e16076","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-08 17:13:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bcf3fec4a61d9a4c82bd72170d1f34ccd6a31096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 18:54:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/844056d/ : SUCCESS in 4m 29s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/95d9251/ : SUCCESS in 4m 23s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/1fb24ef/ : SUCCESS in 8m 23s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/81fb92e/ : SUCCESS in 4m 13s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/6ef5052/ : SUCCESS in 4m 47s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/a9ba1f6/ : SUCCESS in 4m 52s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/eb9cea6/ : SUCCESS in 41m 24s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/bfef933/ : SUCCESS in 4m 42s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/682027e/ : SUCCESS in 32m 03s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/6cf9416/ : SUCCESS in 34m 39s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/a99d255/ : SUCCESS in 1h 04m 26s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/7e13310/ : SUCCESS in 54m 04s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/cf63aae/ : SUCCESS in 47m 43s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/e0f4671/ : SUCCESS in 36m 35s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/99dbe99/ : SUCCESS in 42m 18s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/2ab62e4/ : SUCCESS in 1h 35m 50s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/ca680c4/ : SUCCESS in 35m 27s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/8e304ce/ : SUCCESS in 37m 18s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/5645ca9/ : SUCCESS in 1h 21m 37s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/a697150/ : SUCCESS in 43m 09s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/88769bb/ : SUCCESS in 1h 18m 26s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/1b257b6/ : SUCCESS in 39m 15s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/7b4bb0b/ : SUCCESS in 34m 50s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/f93d8f0/ : SUCCESS in 36m 49s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/7a44efc/ : FAILURE in 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ef3a518d34de00c4674d0802f94eeb1bfc14703","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-10 10:17:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edb59de29f69b2633e13da2837c9694eea8184c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 12:01:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/95f5f84/ : SUCCESS in 4m 30s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/4179fdb/ : SUCCESS in 5m 13s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/7a0cb65/ : SUCCESS in 5m 56s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/bb43936/ : SUCCESS in 5m 13s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/64e540a/ : SUCCESS in 5m 39s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/645acc0/ : SUCCESS in 4m 58s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/ed7007c/ : SUCCESS in 43m 34s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/c278438/ : SUCCESS in 4m 44s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/057d473/ : SUCCESS in 32m 48s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/3edd75b/ : SUCCESS in 33m 27s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/d85de3b/ : SUCCESS in 1h 10m 03s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/2ede9e8/ : SUCCESS in 53m 34s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/24bf5c4/ : SUCCESS in 48m 36s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/5a43dcf/ : SUCCESS in 35m 13s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/7b9da56/ : SUCCESS in 42m 23s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/6e600b5/ : SUCCESS in 1h 36m 30s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/cc3a143/ : SUCCESS in 34m 51s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/ae7fc93/ : SUCCESS in 37m 34s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/cc519d0/ : SUCCESS in 1h 24m 20s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/1c207e3/ : SUCCESS in 38m 20s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/ee3ff5e/ : SUCCESS in 1h 12m 35s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/5944880/ : SUCCESS in 39m 59s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/41a28ce/ : FAILURE in 46m 43s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/322e9fb/ : SUCCESS in 32m 20s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/df03d2c/ : FAILURE in 36m 50s","accounts_in_message":[],"_revision_number":1},{"id":"469f02a8c22a0742cb280520b103ccbda03f5909","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-14 00:15:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c2943afa6a7f2ec15356118da04097ea49c70821","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-05-14 01:23:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d966be31db238ac27e7285abd2c269be1ba0b425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 01:33:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/1/check/rally-tox-pep8/4f5c81c/ : SUCCESS in 4m 40s\n- rally-tox-py27 http://logs.openstack.org/06/657306/1/check/rally-tox-py27/d188ef6/ : SUCCESS in 4m 21s\n- rally-tox-py35 http://logs.openstack.org/06/657306/1/check/rally-tox-py35/3bd8f80/ : SUCCESS in 7m 58s\n- rally-tox-py36 http://logs.openstack.org/06/657306/1/check/rally-tox-py36/0d9492c/ : SUCCESS in 5m 03s\n- rally-tox-py37 http://logs.openstack.org/06/657306/1/check/rally-tox-py37/d6aa4ed/ : SUCCESS in 5m 25s\n- rally-tox-cover http://logs.openstack.org/06/657306/1/check/rally-tox-cover/0881a35/ : SUCCESS in 4m 47s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/1/check/rally-dsvm-tox-functional/619c9be/ : SUCCESS in 37m 41s\n- rally-docker-check http://logs.openstack.org/06/657306/1/check/rally-docker-check/8b0d10c/ : SUCCESS in 5m 13s\n- rally-task-simple-job http://logs.openstack.org/06/657306/1/check/rally-task-simple-job/59d816c/ : SUCCESS in 30m 49s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/1/check/rally-task-basic-with-existing-users/2fc36b9/ : SUCCESS in 31m 01s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/1/check/rally-task-barbican/d72fe37/ : SUCCESS in 1h 05m 21s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/1/check/rally-task-cinder/51eb079/ : SUCCESS in 56m 46s\n- rally-task-heat http://logs.openstack.org/06/657306/1/check/rally-task-heat/bae2b96/ : SUCCESS in 54m 50s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/1/check/rally-task-ironic/cf26ba7/ : FAILURE in 53m 04s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/1/check/rally-task-keystone-glance-swift/f902834/ : SUCCESS in 44m 54s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/1/check/rally-task-load-balancing/4cd76bd/ : FAILURE in 34m 34s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/1/check/rally-task-mistral/57732a5/ : SUCCESS in 33m 03s\n- rally-task-murano http://logs.openstack.org/06/657306/1/check/rally-task-murano/8f800d1/ : SUCCESS in 37m 22s\n- rally-task-neutron http://logs.openstack.org/06/657306/1/check/rally-task-neutron/09a5b7b/ : SUCCESS in 1h 09m 29s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/1/check/rally-task-neutron-with-extensions/0971e1c/ : SUCCESS in 40m 39s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/1/check/rally-task-nova/d9f5754/ : SUCCESS in 1h 15m 18s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/1/check/rally-task-telemetry/2b7fe59/ : SUCCESS in 41m 33s\n- rally-task-watcher http://logs.openstack.org/06/657306/1/check/rally-task-watcher/3b45fab/ : SUCCESS in 33m 23s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/1/check/rally-task-zaqar/1f57c4f/ : SUCCESS in 32m 36s\n- rally-verify-tempest http://logs.openstack.org/06/657306/1/check/rally-verify-tempest/d44ee9c/ : POST_FAILURE in 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4d529df1f66e0734f6c59fc41b062d41868a8d05","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-15 01:10:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"702c0f835f51d0651dd274d0fd7ee74ecaf78c16","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-15 01:29:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3fca0850f35e98f7225daebc46efe4c7dd14a3ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 02:59:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/2/check/rally-tox-pep8/0a02846/ : SUCCESS in 4m 11s\n- rally-tox-py27 http://logs.openstack.org/06/657306/2/check/rally-tox-py27/34fab87/ : SUCCESS in 3m 42s\n- rally-tox-py35 http://logs.openstack.org/06/657306/2/check/rally-tox-py35/efac5a1/ : SUCCESS in 7m 59s\n- rally-tox-py36 http://logs.openstack.org/06/657306/2/check/rally-tox-py36/95f8594/ : SUCCESS in 4m 44s\n- rally-tox-py37 http://logs.openstack.org/06/657306/2/check/rally-tox-py37/f1d2e11/ : SUCCESS in 4m 44s\n- rally-tox-cover http://logs.openstack.org/06/657306/2/check/rally-tox-cover/338b8e2/ : SUCCESS in 4m 11s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/2/check/rally-dsvm-tox-functional/7cdedee/ : SUCCESS in 33m 06s\n- rally-docker-check http://logs.openstack.org/06/657306/2/check/rally-docker-check/9f29a80/ : SUCCESS in 5m 18s\n- rally-task-simple-job http://logs.openstack.org/06/657306/2/check/rally-task-simple-job/848469e/ : SUCCESS in 25m 57s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/2/check/rally-task-basic-with-existing-users/703a3d9/ : SUCCESS in 35m 18s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/2/check/rally-task-barbican/75bfa10/ : SUCCESS in 1h 04m 32s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/2/check/rally-task-cinder/7235dac/ : SUCCESS in 1h 05m 50s\n- rally-task-heat http://logs.openstack.org/06/657306/2/check/rally-task-heat/1ec71f5/ : SUCCESS in 48m 56s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/2/check/rally-task-ironic/d0688d6/ : SUCCESS in 47m 14s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/2/check/rally-task-keystone-glance-swift/1737d50/ : SUCCESS in 53m 39s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/2/check/rally-task-load-balancing/10bdc2b/ : FAILURE in 53m 04s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/2/check/rally-task-mistral/cf821eb/ : SUCCESS in 35m 01s\n- rally-task-murano http://logs.openstack.org/06/657306/2/check/rally-task-murano/b8dc49f/ : FAILURE in 35m 13s\n- rally-task-neutron http://logs.openstack.org/06/657306/2/check/rally-task-neutron/4a11575/ : SUCCESS in 1h 28m 16s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/2/check/rally-task-neutron-with-extensions/3a0e56d/ : SUCCESS in 38m 09s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/2/check/rally-task-nova/51a3fc6/ : SUCCESS in 1h 16m 27s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/2/check/rally-task-telemetry/009aa5b/ : SUCCESS in 39m 46s\n- rally-task-watcher http://logs.openstack.org/06/657306/2/check/rally-task-watcher/d0c7419/ : SUCCESS in 32m 13s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/2/check/rally-task-zaqar/3fb8820/ : SUCCESS in 36m 09s\n- rally-verify-tempest http://logs.openstack.org/06/657306/2/check/rally-verify-tempest/01a31a1/ : POST_FAILURE in 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b4bcbd3295a62516aaefedf3b39fc15702779a18","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-15 03:47:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1abb24b8a419f32b923536597a179d2f59bb1a42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 05:23:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/3/check/rally-tox-pep8/c946643/ : SUCCESS in 4m 59s\n- rally-tox-py27 http://logs.openstack.org/06/657306/3/check/rally-tox-py27/3e2826c/ : SUCCESS in 5m 07s\n- rally-tox-py35 http://logs.openstack.org/06/657306/3/check/rally-tox-py35/b762801/ : SUCCESS in 5m 53s\n- rally-tox-py36 http://logs.openstack.org/06/657306/3/check/rally-tox-py36/ac4d3bb/ : SUCCESS in 4m 05s\n- rally-tox-py37 http://logs.openstack.org/06/657306/3/check/rally-tox-py37/0c3cdba/ : SUCCESS in 6m 12s\n- rally-tox-cover http://logs.openstack.org/06/657306/3/check/rally-tox-cover/d791079/ : SUCCESS in 4m 17s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/3/check/rally-dsvm-tox-functional/0a059a1/ : SUCCESS in 43m 46s\n- rally-docker-check http://logs.openstack.org/06/657306/3/check/rally-docker-check/8150654/ : SUCCESS in 5m 16s\n- rally-task-simple-job http://logs.openstack.org/06/657306/3/check/rally-task-simple-job/bde1944/ : SUCCESS in 25m 18s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/3/check/rally-task-basic-with-existing-users/f422712/ : SUCCESS in 27m 41s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/3/check/rally-task-barbican/45fce29/ : SUCCESS in 1h 01m 58s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/3/check/rally-task-cinder/4b17a3d/ : SUCCESS in 49m 10s\n- rally-task-heat http://logs.openstack.org/06/657306/3/check/rally-task-heat/53d6e84/ : SUCCESS in 56m 05s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/3/check/rally-task-ironic/40446bd/ : SUCCESS in 29m 38s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/3/check/rally-task-keystone-glance-swift/8353d11/ : SUCCESS in 52m 56s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/3/check/rally-task-load-balancing/53aa56d/ : FAILURE in 41m 02s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/3/check/rally-task-mistral/271ccc3/ : SUCCESS in 42m 49s\n- rally-task-murano http://logs.openstack.org/06/657306/3/check/rally-task-murano/dfacdf8/ : SUCCESS in 39m 49s\n- rally-task-neutron http://logs.openstack.org/06/657306/3/check/rally-task-neutron/816d8ab/ : SUCCESS in 1h 32m 50s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/3/check/rally-task-neutron-with-extensions/d4f0ce3/ : SUCCESS in 58m 26s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/3/check/rally-task-nova/75a2044/ : SUCCESS in 1h 24m 12s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/3/check/rally-task-telemetry/002cb16/ : SUCCESS in 59m 13s\n- rally-task-watcher http://logs.openstack.org/06/657306/3/check/rally-task-watcher/b71c04a/ : SUCCESS in 49m 17s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/3/check/rally-task-zaqar/15e5333/ : SUCCESS in 34m 59s\n- rally-verify-tempest http://logs.openstack.org/06/657306/3/check/rally-verify-tempest/685521d/ : POST_FAILURE in 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a88a4fe1f1f329eececa0ad738b1e5a5a6b6ba1c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-15 05:28:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebe53657325cd069bb07287ba662042e37959ef2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 06:55:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/4/check/rally-tox-pep8/963a4f0/ : SUCCESS in 4m 12s\n- rally-tox-py27 http://logs.openstack.org/06/657306/4/check/rally-tox-py27/0085e88/ : SUCCESS in 4m 53s\n- rally-tox-py35 http://logs.openstack.org/06/657306/4/check/rally-tox-py35/145a7a0/ : SUCCESS in 14m 37s\n- rally-tox-py36 http://logs.openstack.org/06/657306/4/check/rally-tox-py36/f2b5a2a/ : SUCCESS in 4m 20s\n- rally-tox-py37 http://logs.openstack.org/06/657306/4/check/rally-tox-py37/4cbbf2c/ : SUCCESS in 7m 04s\n- rally-tox-cover http://logs.openstack.org/06/657306/4/check/rally-tox-cover/a577877/ : SUCCESS in 4m 44s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/4/check/rally-dsvm-tox-functional/10b9a64/ : SUCCESS in 43m 28s\n- rally-docker-check http://logs.openstack.org/06/657306/4/check/rally-docker-check/5af070f/ : SUCCESS in 4m 53s\n- rally-task-simple-job http://logs.openstack.org/06/657306/4/check/rally-task-simple-job/56f9953/ : SUCCESS in 38m 30s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/4/check/rally-task-basic-with-existing-users/a29d817/ : SUCCESS in 33m 27s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/4/check/rally-task-barbican/4c8c8e8/ : SUCCESS in 1h 05m 05s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/4/check/rally-task-cinder/8415c64/ : SUCCESS in 57m 47s\n- rally-task-heat http://logs.openstack.org/06/657306/4/check/rally-task-heat/5088553/ : SUCCESS in 47m 28s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/4/check/rally-task-ironic/e88c670/ : SUCCESS in 36m 29s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/4/check/rally-task-keystone-glance-swift/1cd587f/ : SUCCESS in 46m 53s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/4/check/rally-task-load-balancing/57bdb91/ : FAILURE in 36m 54s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/4/check/rally-task-mistral/d285a71/ : SUCCESS in 31m 56s\n- rally-task-murano http://logs.openstack.org/06/657306/4/check/rally-task-murano/beb9e28/ : SUCCESS in 37m 19s\n- rally-task-neutron http://logs.openstack.org/06/657306/4/check/rally-task-neutron/97e411d/ : SUCCESS in 1h 22m 09s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/4/check/rally-task-neutron-with-extensions/3547a5a/ : SUCCESS in 39m 44s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/4/check/rally-task-nova/7984652/ : SUCCESS in 1h 06m 33s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/4/check/rally-task-telemetry/9f401e0/ : SUCCESS in 39m 07s\n- rally-task-watcher http://logs.openstack.org/06/657306/4/check/rally-task-watcher/2239609/ : SUCCESS in 33m 47s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/4/check/rally-task-zaqar/efe7039/ : SUCCESS in 34m 55s\n- rally-verify-tempest http://logs.openstack.org/06/657306/4/check/rally-verify-tempest/4247f7a/ : POST_FAILURE in 33m 21s","accounts_in_message":[],"_revision_number":4},{"id":"37973a5c506ec0ca712c32e62e2dea6846a998e8","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-16 00:24:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e03238a81a3edb4831d4e702ffc03be32a6d1e48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 01:50:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/4/check/rally-tox-pep8/94fe5b0/ : SUCCESS in 4m 03s\n- rally-tox-py27 http://logs.openstack.org/06/657306/4/check/rally-tox-py27/e778cae/ : SUCCESS in 4m 37s\n- rally-tox-py35 http://logs.openstack.org/06/657306/4/check/rally-tox-py35/36ff661/ : SUCCESS in 6m 00s\n- rally-tox-py36 http://logs.openstack.org/06/657306/4/check/rally-tox-py36/298277f/ : SUCCESS in 4m 07s\n- rally-tox-py37 http://logs.openstack.org/06/657306/4/check/rally-tox-py37/1e04c51/ : SUCCESS in 4m 50s\n- rally-tox-cover http://logs.openstack.org/06/657306/4/check/rally-tox-cover/cd868b5/ : SUCCESS in 4m 21s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/4/check/rally-dsvm-tox-functional/50893f7/ : SUCCESS in 37m 48s\n- rally-docker-check http://logs.openstack.org/06/657306/4/check/rally-docker-check/650b35d/ : SUCCESS in 5m 21s\n- rally-task-simple-job http://logs.openstack.org/06/657306/4/check/rally-task-simple-job/8bce0cc/ : SUCCESS in 30m 13s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/4/check/rally-task-basic-with-existing-users/6d30f26/ : SUCCESS in 32m 49s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/4/check/rally-task-barbican/fa66b5a/ : SUCCESS in 1h 04m 09s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/4/check/rally-task-cinder/e379143/ : SUCCESS in 52m 05s\n- rally-task-heat http://logs.openstack.org/06/657306/4/check/rally-task-heat/03ecd92/ : SUCCESS in 48m 06s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/4/check/rally-task-ironic/75e9dd3/ : SUCCESS in 33m 43s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/4/check/rally-task-keystone-glance-swift/8dc567d/ : SUCCESS in 43m 26s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/4/check/rally-task-load-balancing/47df4fb/ : FAILURE in 40m 41s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/4/check/rally-task-mistral/87fb6cb/ : SUCCESS in 33m 48s\n- rally-task-murano http://logs.openstack.org/06/657306/4/check/rally-task-murano/9003ec5/ : SUCCESS in 37m 24s\n- rally-task-neutron http://logs.openstack.org/06/657306/4/check/rally-task-neutron/f5c97a1/ : SUCCESS in 1h 24m 27s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/4/check/rally-task-neutron-with-extensions/a0e7bf9/ : SUCCESS in 41m 29s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/4/check/rally-task-nova/d4e0f12/ : SUCCESS in 1h 15m 00s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/4/check/rally-task-telemetry/81eb34d/ : SUCCESS in 41m 25s\n- rally-task-watcher http://logs.openstack.org/06/657306/4/check/rally-task-watcher/e5324a1/ : SUCCESS in 32m 06s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/4/check/rally-task-zaqar/0c5b407/ : SUCCESS in 32m 38s\n- rally-verify-tempest http://logs.openstack.org/06/657306/4/check/rally-verify-tempest/c94c4a1/ : POST_FAILURE in 28m 33s","accounts_in_message":[],"_revision_number":4},{"id":"77b19e7e0d18811278aa6352df73b0347ac15ea7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 06:15:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"60589e8c815a58b3b667222c7f513c41fc25b2e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 07:46:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/5/check/rally-tox-pep8/27f5118/ : SUCCESS in 4m 52s\n- rally-tox-py27 http://logs.openstack.org/06/657306/5/check/rally-tox-py27/311cb49/ : SUCCESS in 4m 27s\n- rally-tox-py35 http://logs.openstack.org/06/657306/5/check/rally-tox-py35/df6d680/ : SUCCESS in 8m 13s\n- rally-tox-py36 http://logs.openstack.org/06/657306/5/check/rally-tox-py36/29cd347/ : SUCCESS in 4m 02s\n- rally-tox-py37 http://logs.openstack.org/06/657306/5/check/rally-tox-py37/f18d060/ : SUCCESS in 5m 07s\n- rally-tox-cover http://logs.openstack.org/06/657306/5/check/rally-tox-cover/1f704b4/ : SUCCESS in 7m 52s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/5/check/rally-dsvm-tox-functional/c411f57/ : SUCCESS in 41m 00s\n- rally-docker-check http://logs.openstack.org/06/657306/5/check/rally-docker-check/e0a9912/ : SUCCESS in 7m 06s\n- rally-task-simple-job http://logs.openstack.org/06/657306/5/check/rally-task-simple-job/21fb798/ : SUCCESS in 24m 35s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/5/check/rally-task-basic-with-existing-users/62f1b2e/ : SUCCESS in 31m 33s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/5/check/rally-task-barbican/b5cf7cb/ : SUCCESS in 1h 02m 31s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/5/check/rally-task-cinder/2daa81f/ : SUCCESS in 53m 51s\n- rally-task-heat http://logs.openstack.org/06/657306/5/check/rally-task-heat/8abca4c/ : SUCCESS in 52m 05s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/5/check/rally-task-ironic/52688bf/ : SUCCESS in 37m 58s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/5/check/rally-task-keystone-glance-swift/a4b5943/ : SUCCESS in 54m 26s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/5/check/rally-task-load-balancing/2f79f72/ : FAILURE in 49m 38s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/5/check/rally-task-mistral/0f27b97/ : SUCCESS in 33m 11s\n- rally-task-murano http://logs.openstack.org/06/657306/5/check/rally-task-murano/c0c6cb7/ : SUCCESS in 38m 02s\n- rally-task-neutron http://logs.openstack.org/06/657306/5/check/rally-task-neutron/4292769/ : SUCCESS in 1h 23m 00s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/5/check/rally-task-neutron-with-extensions/9d88a34/ : SUCCESS in 44m 23s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/5/check/rally-task-nova/5ea62e7/ : SUCCESS in 1h 10m 26s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/5/check/rally-task-telemetry/2709380/ : SUCCESS in 34m 30s\n- rally-task-watcher http://logs.openstack.org/06/657306/5/check/rally-task-watcher/822a5ae/ : SUCCESS in 28m 21s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/5/check/rally-task-zaqar/3efbbd1/ : SUCCESS in 33m 25s\n- rally-verify-tempest http://logs.openstack.org/06/657306/5/check/rally-verify-tempest/1d20091/ : POST_FAILURE in 29m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c43d5d9f5a39b38689c50f870aab39c584c7daae","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 08:03:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9ba65bfaa3322d5ea78139e9f1a2c2d691d1411d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 09:32:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/6/check/rally-tox-pep8/bb54dae/ : SUCCESS in 4m 47s\n- rally-tox-py27 http://logs.openstack.org/06/657306/6/check/rally-tox-py27/c2e2330/ : SUCCESS in 4m 02s\n- rally-tox-py35 http://logs.openstack.org/06/657306/6/check/rally-tox-py35/f2077f1/ : SUCCESS in 9m 13s\n- rally-tox-py36 http://logs.openstack.org/06/657306/6/check/rally-tox-py36/7eda70a/ : SUCCESS in 4m 15s\n- rally-tox-py37 http://logs.openstack.org/06/657306/6/check/rally-tox-py37/385a6c9/ : SUCCESS in 5m 59s\n- rally-tox-cover http://logs.openstack.org/06/657306/6/check/rally-tox-cover/719e57f/ : SUCCESS in 5m 36s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/6/check/rally-dsvm-tox-functional/9e7f655/ : SUCCESS in 31m 13s\n- rally-docker-check http://logs.openstack.org/06/657306/6/check/rally-docker-check/aa1ef2f/ : SUCCESS in 9m 14s\n- rally-task-simple-job http://logs.openstack.org/06/657306/6/check/rally-task-simple-job/34c5487/ : SUCCESS in 42m 37s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/6/check/rally-task-basic-with-existing-users/d0309c7/ : SUCCESS in 30m 49s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/6/check/rally-task-barbican/d2a8f2b/ : SUCCESS in 1h 01m 33s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/6/check/rally-task-cinder/b9f5ac3/ : SUCCESS in 54m 41s\n- rally-task-heat http://logs.openstack.org/06/657306/6/check/rally-task-heat/598b626/ : SUCCESS in 48m 15s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/6/check/rally-task-ironic/ce21daf/ : SUCCESS in 33m 25s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/6/check/rally-task-keystone-glance-swift/d391e86/ : SUCCESS in 41m 54s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/6/check/rally-task-load-balancing/13c3a9c/ : FAILURE in 39m 53s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/6/check/rally-task-mistral/b085a3b/ : SUCCESS in 32m 47s\n- rally-task-murano http://logs.openstack.org/06/657306/6/check/rally-task-murano/4fecd1b/ : FAILURE in 32m 34s\n- rally-task-neutron http://logs.openstack.org/06/657306/6/check/rally-task-neutron/93c33e9/ : SUCCESS in 1h 24m 13s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/6/check/rally-task-neutron-with-extensions/1eb1073/ : SUCCESS in 38m 54s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/6/check/rally-task-nova/5361276/ : SUCCESS in 1h 13m 16s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/6/check/rally-task-telemetry/3b56851/ : SUCCESS in 36m 45s\n- rally-task-watcher http://logs.openstack.org/06/657306/6/check/rally-task-watcher/bd331f7/ : SUCCESS in 32m 11s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/6/check/rally-task-zaqar/cec4382/ : SUCCESS in 28m 20s\n- rally-verify-tempest http://logs.openstack.org/06/657306/6/check/rally-verify-tempest/c50a30c/ : FAILURE in 31m 43s","accounts_in_message":[],"_revision_number":6},{"id":"e209edafced57989f78648d185f21601b120f3f4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 16:27:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"471d64dcb2d47a9a41f0eb673e79805498d4bab9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 17:28:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"665a7de7a6d9f0ee3f2ce22b607bbb1fa7cfb792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:10:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/8/check/rally-tox-pep8/ccd4b66/ : SUCCESS in 4m 28s\n- rally-tox-py27 http://logs.openstack.org/06/657306/8/check/rally-tox-py27/be48c5b/ : SUCCESS in 3m 46s\n- rally-tox-py35 http://logs.openstack.org/06/657306/8/check/rally-tox-py35/b195eb3/ : SUCCESS in 8m 04s\n- rally-tox-py36 http://logs.openstack.org/06/657306/8/check/rally-tox-py36/4fcee81/ : SUCCESS in 4m 06s\n- rally-tox-py37 http://logs.openstack.org/06/657306/8/check/rally-tox-py37/642df38/ : SUCCESS in 7m 19s\n- rally-tox-cover http://logs.openstack.org/06/657306/8/check/rally-tox-cover/5774391/ : SUCCESS in 4m 45s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/8/check/rally-dsvm-tox-functional/05a709c/ : SUCCESS in 40m 36s\n- rally-docker-check http://logs.openstack.org/06/657306/8/check/rally-docker-check/9979ca6/ : SUCCESS in 6m 27s\n- rally-task-simple-job http://logs.openstack.org/06/657306/8/check/rally-task-simple-job/516f603/ : SUCCESS in 32m 34s\n- rally-task-basic-with-existing-users http://logs.openstack.org/06/657306/8/check/rally-task-basic-with-existing-users/4601fe3/ : SUCCESS in 34m 27s (non-voting)\n- rally-task-barbican http://logs.openstack.org/06/657306/8/check/rally-task-barbican/90e9222/ : SUCCESS in 1h 06m 34s (non-voting)\n- rally-task-cinder http://logs.openstack.org/06/657306/8/check/rally-task-cinder/74459a5/ : SUCCESS in 51m 04s\n- rally-task-heat http://logs.openstack.org/06/657306/8/check/rally-task-heat/285cea7/ : SUCCESS in 45m 35s (non-voting)\n- rally-task-ironic http://logs.openstack.org/06/657306/8/check/rally-task-ironic/9c8986b/ : SUCCESS in 40m 40s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/8/check/rally-task-keystone-glance-swift/af71250/ : SUCCESS in 40m 48s\n- rally-task-load-balancing http://logs.openstack.org/06/657306/8/check/rally-task-load-balancing/224eedf/ : FAILURE in 44m 16s (non-voting)\n- rally-task-mistral http://logs.openstack.org/06/657306/8/check/rally-task-mistral/4c75610/ : SUCCESS in 36m 47s\n- rally-task-murano http://logs.openstack.org/06/657306/8/check/rally-task-murano/27ad670/ : SUCCESS in 38m 38s\n- rally-task-neutron http://logs.openstack.org/06/657306/8/check/rally-task-neutron/3df56bf/ : SUCCESS in 1h 37m 01s\n- rally-task-neutron-with-extensions http://logs.openstack.org/06/657306/8/check/rally-task-neutron-with-extensions/02665b7/ : SUCCESS in 39m 24s (non-voting)\n- rally-task-nova http://logs.openstack.org/06/657306/8/check/rally-task-nova/bc08051/ : SUCCESS in 1h 18m 20s (non-voting)\n- rally-task-telemetry http://logs.openstack.org/06/657306/8/check/rally-task-telemetry/9b86109/ : SUCCESS in 38m 18s\n- rally-task-watcher http://logs.openstack.org/06/657306/8/check/rally-task-watcher/4bfa9a8/ : SUCCESS in 35m 29s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/06/657306/8/check/rally-task-zaqar/7ff1712/ : SUCCESS in 35m 35s\n- rally-verify-tempest http://logs.openstack.org/06/657306/8/check/rally-verify-tempest/e4ceaf0/ : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":8},{"id":"98979c66693f6cdd396c8ad28d420d25986468d8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 19:50:04.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nfinally","accounts_in_message":[],"_revision_number":8},{"id":"ef69e95c079ae0cc5ff76728208ecaaec4f5b461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:50:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7499dfad8baf0f5243beefcf1389f9a988e30c69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:24:54.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- rally-tox-pep8 http://logs.openstack.org/06/657306/8/gate/rally-tox-pep8/145641c/ : SUCCESS in 4m 48s\n- rally-tox-py27 http://logs.openstack.org/06/657306/8/gate/rally-tox-py27/92a522d/ : SUCCESS in 4m 44s\n- rally-tox-py35 http://logs.openstack.org/06/657306/8/gate/rally-tox-py35/c2fec4e/ : SUCCESS in 7m 31s\n- rally-tox-py36 http://logs.openstack.org/06/657306/8/gate/rally-tox-py36/f1803d0/ : SUCCESS in 4m 17s\n- rally-tox-py37 http://logs.openstack.org/06/657306/8/gate/rally-tox-py37/7c39f34/ : SUCCESS in 5m 24s\n- rally-tox-cover http://logs.openstack.org/06/657306/8/gate/rally-tox-cover/9a3745e/ : SUCCESS in 5m 28s\n- rally-dsvm-tox-functional http://logs.openstack.org/06/657306/8/gate/rally-dsvm-tox-functional/db53efd/ : SUCCESS in 38m 05s\n- rally-docker-check http://logs.openstack.org/06/657306/8/gate/rally-docker-check/d5c8daa/ : SUCCESS in 4m 30s\n- rally-task-simple-job http://logs.openstack.org/06/657306/8/gate/rally-task-simple-job/c9e7c8e/ : SUCCESS in 30m 51s\n- rally-task-cinder http://logs.openstack.org/06/657306/8/gate/rally-task-cinder/53d7508/ : SUCCESS in 1h 00m 45s\n- rally-task-ironic http://logs.openstack.org/06/657306/8/gate/rally-task-ironic/a078e64/ : SUCCESS in 38m 40s\n- rally-task-keystone-glance-swift http://logs.openstack.org/06/657306/8/gate/rally-task-keystone-glance-swift/0fa0327/ : SUCCESS in 41m 34s\n- rally-task-mistral http://logs.openstack.org/06/657306/8/gate/rally-task-mistral/215716c/ : SUCCESS in 41m 02s\n- rally-task-murano http://logs.openstack.org/06/657306/8/gate/rally-task-murano/128efc5/ : SUCCESS in 37m 43s\n- rally-task-neutron http://logs.openstack.org/06/657306/8/gate/rally-task-neutron/10780c6/ : SUCCESS in 1h 30m 20s\n- rally-task-telemetry http://logs.openstack.org/06/657306/8/gate/rally-task-telemetry/ce37201/ : SUCCESS in 39m 38s\n- rally-task-zaqar http://logs.openstack.org/06/657306/8/gate/rally-task-zaqar/9195ccf/ : SUCCESS in 33m 13s\n- rally-verify-tempest http://logs.openstack.org/06/657306/8/gate/rally-verify-tempest/4081a92/ : SUCCESS in 35m 42s","accounts_in_message":[],"_revision_number":8},{"id":"ae51d58cebbc69b7dabe848ebd1990d8f22a7709","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:24:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a539f306c15178e1f69679e46bfb5103173f283d","revisions":{"5c67c4d9fc0bc4a0341f8d17efdd38ff46f7db0b":{"kind":"REWORK","_number":1,"created":"2019-05-06 03:03:25.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/06/657306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/1"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:24:04.000000000","tz":480},"subject":"Fix unit tests issue","message":"Fix unit tests issue\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5c67c4d9fc0bc4a0341f8d17efdd38ff46f7db0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5c67c4d9fc0bc4a0341f8d17efdd38ff46f7db0b"}]},"branch":"refs/heads/master"},"8ea947d4c33585d988e318c44b2908a6831fcf7f":{"kind":"REWORK","_number":2,"created":"2019-05-15 01:29:25.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/2"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:28:50.000000000","tz":-420},"subject":"Fix various testing","message":"Fix various testing\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. move rally-verify job under python3 env\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/8ea947d4c33585d988e318c44b2908a6831fcf7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/8ea947d4c33585d988e318c44b2908a6831fcf7f"}]},"branch":"refs/heads/master"},"643b8ac325469922ceb0027c55028d8a1c16fcf1":{"kind":"REWORK","_number":3,"created":"2019-05-15 03:47:50.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/3"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 03:47:44.000000000","tz":-420},"subject":"Fix various testing","message":"Fix various testing\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. move rally-verify job under python3 env\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/643b8ac325469922ceb0027c55028d8a1c16fcf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/643b8ac325469922ceb0027c55028d8a1c16fcf1"}]},"branch":"refs/heads/master"},"3cd496a316c982a5cfa8844e86c14e84cd7c170f":{"kind":"REWORK","_number":4,"created":"2019-05-15 05:28:23.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/4"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 05:28:19.000000000","tz":-420},"subject":"Fix various testing","message":"Fix various testing\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. move rally-verify job under python3 env\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/3cd496a316c982a5cfa8844e86c14e84cd7c170f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/3cd496a316c982a5cfa8844e86c14e84cd7c170f"}]},"branch":"refs/heads/master"},"b5b488bedf5dd39ddca9c5215ef379882a1fbf1a":{"kind":"REWORK","_number":5,"created":"2019-05-16 06:15:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/5"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 06:15:42.000000000","tz":-420},"subject":"Fix various testing","message":"Fix various testing\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. move rally-verify job under python3 env\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/b5b488bedf5dd39ddca9c5215ef379882a1fbf1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/b5b488bedf5dd39ddca9c5215ef379882a1fbf1a"}]},"branch":"refs/heads/master"},"aa383f7e94282e9e1a7db84dd0c93ab35064cbdd":{"kind":"REWORK","_number":6,"created":"2019-05-16 08:03:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/6"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 08:03:29.000000000","tz":-420},"subject":"Fix various testing","message":"Fix various testing\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. move rally-verify job under python3 env\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/aa383f7e94282e9e1a7db84dd0c93ab35064cbdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/aa383f7e94282e9e1a7db84dd0c93ab35064cbdd"}]},"branch":"refs/heads/master"},"d6f7cbedf8874de08035b86adb1670d04c875625":{"kind":"REWORK","_number":7,"created":"2019-05-16 16:27:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/7"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 16:26:02.000000000","tz":-420},"subject":"Fix various issues","message":"Fix various issues\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. Move rally-verify job under python3 env\n4. Switch tempest verifier default source to opendev.org\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/d6f7cbedf8874de08035b86adb1670d04c875625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/d6f7cbedf8874de08035b86adb1670d04c875625"}]},"branch":"refs/heads/master"},"a539f306c15178e1f69679e46bfb5103173f283d":{"kind":"REWORK","_number":8,"created":"2019-05-16 17:28:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/06/657306/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/06/657306/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/06/657306/8"}}},"commit":{"parents":[{"commit":"945ac47507cbae42fea18a07c8193618fab664f9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/945ac47507cbae42fea18a07c8193618fab664f9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-05-06 03:21:51.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 17:28:01.000000000","tz":-420},"subject":"Fix various issues","message":"Fix various issues\n\n1. Specify Sphinx version while using Python2.7\n2. Fix mock verification in test_check_health_with_api_info\n3. Move rally-verify job under python3 env\n4. Switch tempest verifier default source to opendev.org\n\nChange-Id: I8f92ef87722d057514d0906d068ddaadd88c8758\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/a539f306c15178e1f69679e46bfb5103173f283d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/a539f306c15178e1f69679e46bfb5103173f283d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
