)]}'
{"id":"openstack%2Frally-openstack~670140","triplet_id":"openstack%2Frally-openstack~master~I43e6c253b11128a7233fcfafeab3e0175c553894","project":"openstack/rally-openstack","branch":"master","topic":"ci_jobs","hashtags":[],"change_id":"I43e6c253b11128a7233fcfafeab3e0175c553894","subject":"[ci] Re-order jobs","status":"MERGED","created":"2019-07-10 17:14:56.000000000","updated":"2019-07-10 20:50:55.000000000","submitted":"2019-07-10 20:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670140-1562791855932-9c139318","meta_rev_id":"e7b05b0d30168848b99671d8a233e090b61bf883","_number":670140,"virtual_id_number":670140,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-07-10 20:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-07-10 17:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-07-10 17:15:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 20:50:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"68396dce82c2d1e5d86c7c8330916b48f6ba370d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-07-10 17:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e82b3fcf409d23510d05bc33b758a1c382a435e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-07-10 17:15:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f1ad424433a84b6c287faaa2949f69d2b7f073a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- rally-tox-cover http://logs.openstack.org/40/670140/1/check/rally-tox-cover/435e8f3/ : SUCCESS in 5m 08s\n- rally-tox-pep8 http://logs.openstack.org/40/670140/1/check/rally-tox-pep8/b315393/ : SUCCESS in 5m 42s\n- rally-tox-py27 http://logs.openstack.org/40/670140/1/check/rally-tox-py27/c83cd80/ : SUCCESS in 5m 34s\n- rally-tox-py35 http://logs.openstack.org/40/670140/1/check/rally-tox-py35/dc5abac/ : SUCCESS in 7m 08s\n- rally-tox-py36 http://logs.openstack.org/40/670140/1/check/rally-tox-py36/bdf1c0d/ : SUCCESS in 5m 34s\n- rally-tox-py37 http://logs.openstack.org/40/670140/1/check/rally-tox-py37/c488dff/ : SUCCESS in 6m 27s\n- rally-dsvm-tox-functional http://logs.openstack.org/40/670140/1/check/rally-dsvm-tox-functional/da5d020/ : SUCCESS in 38m 26s\n- rally-docker-check http://logs.openstack.org/40/670140/1/check/rally-docker-check/c522e5f/ : SUCCESS in 5m 11s\n- rally-task-basic-with-existing-users http://logs.openstack.org/40/670140/1/check/rally-task-basic-with-existing-users/9595f8e/ : SUCCESS in 33m 20s (non-voting)\n- rally-task-simple-job http://logs.openstack.org/40/670140/1/check/rally-task-simple-job/08642da/ : SUCCESS in 33m 46s\n- rally-task-barbican http://logs.openstack.org/40/670140/1/check/rally-task-barbican/bb42d67/ : SUCCESS in 32m 25s (non-voting)\n- rally-task-cinder http://logs.openstack.org/40/670140/1/check/rally-task-cinder/6cc54b4/ : SUCCESS in 53m 13s\n- rally-task-heat http://logs.openstack.org/40/670140/1/check/rally-task-heat/53ede0d/ : SUCCESS in 47m 02s (non-voting)\n- rally-task-ironic http://logs.openstack.org/40/670140/1/check/rally-task-ironic/b079fc1/ : SUCCESS in 34m 54s\n- rally-task-keystone-glance-swift http://logs.openstack.org/40/670140/1/check/rally-task-keystone-glance-swift/3094457/ : SUCCESS in 50m 42s\n- rally-task-mistral http://logs.openstack.org/40/670140/1/check/rally-task-mistral/f636a19/ : SUCCESS in 35m 30s\n- rally-task-murano http://logs.openstack.org/40/670140/1/check/rally-task-murano/a221450/ : SUCCESS in 37m 26s\n- rally-task-neutron http://logs.openstack.org/40/670140/1/check/rally-task-neutron/ffb96ca/ : SUCCESS in 1h 35m 55s\n- rally-task-neutron-with-extensions http://logs.openstack.org/40/670140/1/check/rally-task-neutron-with-extensions/a827370/ : SUCCESS in 40m 35s (non-voting)\n- rally-task-nova http://logs.openstack.org/40/670140/1/check/rally-task-nova/45693b2/ : SUCCESS in 1h 17m 35s (non-voting)\n- rally-task-octavia http://logs.openstack.org/40/670140/1/check/rally-task-octavia/8194525/ : SUCCESS in 1h 40m 17s\n- rally-task-telemetry http://logs.openstack.org/40/670140/1/check/rally-task-telemetry/9f8fe14/ : SUCCESS in 44m 25s\n- rally-task-watcher http://logs.openstack.org/40/670140/1/check/rally-task-watcher/b425a2d/ : SUCCESS in 38m 15s (non-voting)\n- rally-task-zaqar http://logs.openstack.org/40/670140/1/check/rally-task-zaqar/1669077/ : SUCCESS in 44m 43s\n- rally-verify-tempest http://logs.openstack.org/40/670140/1/check/rally-verify-tempest/f78f3d5/ : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"931f9d97466f1f915e758cfb54adb191a0b9b3bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 19:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a0a2f8981c892b9d63cf5e8c62f1f53b6247389","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:50:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"91414f44a356640aab8950ac1456f81c139efec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:50:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- rally-tox-cover http://logs.openstack.org/40/670140/1/gate/rally-tox-cover/5250bcb/ : SUCCESS in 5m 36s\n- rally-tox-pep8 http://logs.openstack.org/40/670140/1/gate/rally-tox-pep8/fc3a05a/ : SUCCESS in 5m 35s\n- rally-tox-py27 http://logs.openstack.org/40/670140/1/gate/rally-tox-py27/a8ed021/ : SUCCESS in 4m 47s\n- rally-tox-py35 http://logs.openstack.org/40/670140/1/gate/rally-tox-py35/160f7d7/ : SUCCESS in 5m 46s\n- rally-tox-py36 http://logs.openstack.org/40/670140/1/gate/rally-tox-py36/5cb2002/ : SUCCESS in 5m 07s\n- rally-tox-py37 http://logs.openstack.org/40/670140/1/gate/rally-tox-py37/4acbeca/ : SUCCESS in 4m 55s\n- rally-dsvm-tox-functional http://logs.openstack.org/40/670140/1/gate/rally-dsvm-tox-functional/c012ef5/ : SUCCESS in 41m 40s\n- rally-docker-check http://logs.openstack.org/40/670140/1/gate/rally-docker-check/30178ec/ : SUCCESS in 4m 57s\n- rally-task-simple-job http://logs.openstack.org/40/670140/1/gate/rally-task-simple-job/4f6c354/ : SUCCESS in 29m 23s\n- rally-task-cinder http://logs.openstack.org/40/670140/1/gate/rally-task-cinder/99c6c84/ : SUCCESS in 54m 35s\n- rally-task-ironic http://logs.openstack.org/40/670140/1/gate/rally-task-ironic/59bcebc/ : SUCCESS in 39m 10s\n- rally-task-keystone-glance-swift http://logs.openstack.org/40/670140/1/gate/rally-task-keystone-glance-swift/5320fac/ : SUCCESS in 42m 48s\n- rally-task-mistral http://logs.openstack.org/40/670140/1/gate/rally-task-mistral/e8a9941/ : SUCCESS in 35m 25s\n- rally-task-murano http://logs.openstack.org/40/670140/1/gate/rally-task-murano/83362ba/ : SUCCESS in 45m 26s\n- rally-task-neutron http://logs.openstack.org/40/670140/1/gate/rally-task-neutron/9c9653b/ : SUCCESS in 1h 36m 00s\n- rally-task-octavia http://logs.openstack.org/40/670140/1/gate/rally-task-octavia/ebcd441/ : SUCCESS in 1h 30m 03s\n- rally-task-telemetry http://logs.openstack.org/40/670140/1/gate/rally-task-telemetry/7424d5e/ : SUCCESS in 43m 00s\n- rally-task-zaqar http://logs.openstack.org/40/670140/1/gate/rally-task-zaqar/d0c5254/ : SUCCESS in 36m 21s\n- rally-verify-tempest http://logs.openstack.org/40/670140/1/gate/rally-verify-tempest/a2fa914/ : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ebcb670a46e0212380f120c0df3fbf2bd823db4","revisions":{"5ebcb670a46e0212380f120c0df3fbf2bd823db4":{"kind":"REWORK","_number":1,"created":"2019-07-10 17:14:56.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/40/670140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/40/670140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/40/670140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/40/670140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/40/670140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/40/670140/1"}}},"commit":{"parents":[{"commit":"f1edeb6c54ffc4ab02e5bd7f3ef77982fd4466d9","subject":"Merge \"Update logging of users context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/f1edeb6c54ffc4ab02e5bd7f3ef77982fd4466d9"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-07-10 17:14:51.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-07-10 17:14:51.000000000","tz":180},"subject":"[ci] Re-order jobs","message":"[ci] Re-order jobs\n\nChange-Id: I43e6c253b11128a7233fcfafeab3e0175c553894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5ebcb670a46e0212380f120c0df3fbf2bd823db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/5ebcb670a46e0212380f120c0df3fbf2bd823db4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
