)]}'
{"id":"openstack%2Frally-openstack~696314","triplet_id":"openstack%2Frally-openstack~master~I4cb59a1c20a16328f8fb442d63ba5ed0f35b851e","project":"openstack/rally-openstack","branch":"master","topic":"fix_ci","hashtags":[],"change_id":"I4cb59a1c20a16328f8fb442d63ba5ed0f35b851e","subject":"Temporary turn off failing tests","status":"MERGED","created":"2019-11-27 14:37:30.000000000","updated":"2019-11-27 18:45:27.000000000","submitted":"2019-11-27 18:45:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696314-1574880327143-fe519b9b","meta_rev_id":"239141141088d3095042584189495e3b35e3f413","_number":696314,"virtual_id_number":696314,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-27 18:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-11-27 16:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-11-27 16:05:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 18:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4319e382625fa297db98fed48edd4b0d28730d5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-27 14:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"804dc525fe3b6708cfe6816ca7d59537f815296d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-27 14:37:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15cf42c983f6b4234dfd61224f72b5c28d87eea2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-27 16:00:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a73943ba4219c9303abd5ed19db49849bb1d836b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-27 16:05:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3f64427522b36a0bd9c0b54ed269cb017b5dc34f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:38:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/eb01d4bb24a94795800f81a8cfd75b11 : SUCCESS in 5m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/36dd8c07e4c140708d071f50a533212b : SUCCESS in 4m 35s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/8a5f31fc41a341eda912cba14a69bf72 : SUCCESS in 6m 13s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/5e40e6651fbb4882b907c87cb3e51d96 : SUCCESS in 5m 08s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a102ba63a603491e922a263d17cfb247 : SUCCESS in 4m 54s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/b13392ddff7e47e4b5a0f057766cf663 : SUCCESS in 6m 43s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/7daf202681d84364a070d067569014c2 : SUCCESS in 45m 44s\n- rally-docker-check https://zuul.opendev.org/t/openstack/build/c5959545cad14dcdb134ae97777154a3 : SUCCESS in 6m 49s\n- rally-task-basic-with-existing-users https://zuul.opendev.org/t/openstack/build/cbb3f5f822564a1082eda5167ee1a034 : SUCCESS in 42m 34s (non-voting)\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/09c12d52e1b2475a96a3365c732f73dc : SUCCESS in 36m 22s\n- rally-task-barbican https://zuul.opendev.org/t/openstack/build/fc623f01b9ba45618d822eede602fe84 : FAILURE in 44m 23s (non-voting)\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/913cd3ef7e9346948dbb364fff316bef : SUCCESS in 56m 04s\n- rally-task-heat https://zuul.opendev.org/t/openstack/build/4bc6b4c20f7145ddad888f7469eb512c : SUCCESS in 46m 32s (non-voting)\n- rally-task-ironic https://zuul.opendev.org/t/openstack/build/5741f22f4a0a43fd96c556eeecdd0858 : SUCCESS in 43m 46s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/c322ae7d3ffd41ffbc020044e581d26a : SUCCESS in 50m 14s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/12903560a6674bdea6bed4469dc3f248 : SUCCESS in 42m 10s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/76e9756245184fe7960b74edd53e07ac : SUCCESS in 42m 46s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/1caba77490db4e8a920470fd77757523 : SUCCESS in 40m 20s\n- rally-task-murano https://zuul.opendev.org/t/openstack/build/2fac40d66fd54ddea9c89f3294e41c1a : FAILURE in 47m 37s (non-voting)\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/2c88a96d92ed4dbbb646b0b36a4cef7d : SUCCESS in 1h 27m 15s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/4f665b1f96e14b2da2983b69a6771cd6 : SUCCESS in 47m 25s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/c8782edbc12d4b009fb0e6db3c4ddc4e : SUCCESS in 1h 00m 20s (non-voting)\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/b63ee20158d246fc8e295d90e0a99292 : SUCCESS in 36m 53s\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/d1889acdf9724e80a477aa5dca7ec263 : SUCCESS in 38m 40s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/3494514d3eb843b493d6a3255267fcea : SUCCESS in 40m 05s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/01a8b6dcdb5044d3bf330656e682daa7 : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1e9710608afee2aaff74b946b0322c709e2a23c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 17:38:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"726cb7f34dd4c550af4258a16cb02a365f3674da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:45:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/62b3ac67b4dc4bbbb48eea2cc88285ee : SUCCESS in 5m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b534416124a4a7d93555443e851d185 : SUCCESS in 6m 55s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/1839fc64e3f44bd087892238e8cdc326 : SUCCESS in 4m 59s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/5340cf32dc0a45dda97531756cb41b3e : SUCCESS in 6m 53s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/120d592a61b24b3bb8f66e2d5a52099b : SUCCESS in 7m 15s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/8af4126fa78d4494adfcc6b19e180f90 : SUCCESS in 6m 46s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/4dfb14b5bc254f0ba66bf9a0d955c052 : SUCCESS in 49m 50s\n- rally-docker-check https://zuul.opendev.org/t/openstack/build/1496d67a381d4b439a65e6b7f2fd5c97 : SUCCESS in 5m 53s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/a74cb7bcf6014809b9111b604fb15481 : SUCCESS in 50m 57s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/50f9abb842b645fbbe278169b43f4017 : SUCCESS in 1h 01m 01s\n- rally-task-ironic https://zuul.opendev.org/t/openstack/build/abb79a55919e4f04a29aab6648f99490 : SUCCESS in 39m 00s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/8d791337eaa6434a885ee90e150dbbab : SUCCESS in 44m 32s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/2de6e031d34846cf843a36fb22c8680d : SUCCESS in 37m 26s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/4993bbd27f4b45fbb4a061670f8bb506 : SUCCESS in 1h 04m 04s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/7eadaed4dc854710b3cb147c18274dc0 : SUCCESS in 46m 20s\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/be2d46d496a74f9c877af786d6fbc053 : SUCCESS in 39m 33s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/1e0138086b634c858ca9cdc6390c12b2 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"72407cb73d7d3d2309f6c09bf6e50ca130127d63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 18:45:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55014fc5bed0c29529a1c3efd3aac2364a4d0b5a","revisions":{"928affc3e1d7de22f908e0e67bee5e4408346999":{"kind":"REWORK","_number":1,"created":"2019-11-27 14:37:30.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/14/696314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/14/696314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/1"}}},"commit":{"parents":[{"commit":"6f426242c7c1fd73d8cd5716a38e7c08f8a8d902","subject":"Merge \"fix number of users for NeutronSubnets.delete_subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/6f426242c7c1fd73d8cd5716a38e7c08f8a8d902"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-27 14:16:27.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-27 14:16:27.000000000","tz":120},"subject":"Temporary turn off failing tests","message":"Temporary turn off failing tests\n\nChange-Id: I4cb59a1c20a16328f8fb442d63ba5ed0f35b851e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/928affc3e1d7de22f908e0e67bee5e4408346999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/928affc3e1d7de22f908e0e67bee5e4408346999"}]},"branch":"refs/heads/master"},"55014fc5bed0c29529a1c3efd3aac2364a4d0b5a":{"kind":"REWORK","_number":2,"created":"2019-11-27 16:00:58.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/14/696314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/14/696314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/14/696314/2"}}},"commit":{"parents":[{"commit":"6f426242c7c1fd73d8cd5716a38e7c08f8a8d902","subject":"Merge \"fix number of users for NeutronSubnets.delete_subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/6f426242c7c1fd73d8cd5716a38e7c08f8a8d902"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-27 14:16:27.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-27 15:41:03.000000000","tz":120},"subject":"Temporary turn off failing tests","message":"Temporary turn off failing tests\n\nSeveral neutron and cinder related issues are blocking our gates.\n\nChange-Id: I4cb59a1c20a16328f8fb442d63ba5ed0f35b851e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55014fc5bed0c29529a1c3efd3aac2364a4d0b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/55014fc5bed0c29529a1c3efd3aac2364a4d0b5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
