)]}'
{"id":"openstack%2Frally-openstack~757545","triplet_id":"openstack%2Frally-openstack~master~Ib3ce073e260c1387f7d26b7d994da2fdc8077e06","project":"openstack/rally-openstack","branch":"master","topic":"ci","hashtags":[],"change_id":"Ib3ce073e260c1387f7d26b7d994da2fdc8077e06","subject":"Fix used python versions for CI","status":"MERGED","created":"2020-10-12 11:50:36.000000000","updated":"2020-10-12 14:45:46.000000000","submitted":"2020-10-12 14:45:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757545-1602513946783-eefd7343","meta_rev_id":"df7efd9d35ee8fe4bcccee330d6b1e04a8aeb641","_number":757545,"virtual_id_number":757545,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 14:45:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-10-12 13:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-10-12 13:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-12 14:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f11e2718a6ae522af653a0914e0fb3ca091c2f84","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-10-12 11:50:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6c4dcf9c3fdf978aa3753128ec46689faf32e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:00:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/983e7c1562d04c53a10d97795ad5466d : SUCCESS in 5m 56s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/64416fe493fe4021b8181c296f52169a : SUCCESS in 4m 53s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f3f4bb23280c47baabf31b58b313999e : SUCCESS in 4m 36s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4641966157f046d1a1a76d01891bf367 : SUCCESS in 4m 24s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/e9deb72474a24f83a9858d815d47301b : SUCCESS in 4m 21s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/b448f3a107bc48a4a5a1931229b66534 : SUCCESS in 48m 44s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/b65db54d31b240c694846f1e170e6e12 : SUCCESS in 8m 04s\n- rally-task-basic-with-existing-users https://zuul.opendev.org/t/openstack/build/41290972a85a4aab832a46224d0bcd83 : SUCCESS in 45m 22s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/cd433c335fdd4b9abe517aa119a01d83 : SUCCESS in 34m 12s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/6d332f3fb2d243adab5c29d7e18d850a : SUCCESS in 55m 12s\n- rally-task-heat https://zuul.opendev.org/t/openstack/build/93bb64e94b8241c0b866a2bdce520e38 : FAILURE in 53m 32s (non-voting)\n- rally-task-ironic https://zuul.opendev.org/t/openstack/build/51e310f014704a0aa8ba4fad9ba54862 : SUCCESS in 46m 40s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/8d51b29c947a4ac3b35893f6b4efb93f : SUCCESS in 48m 29s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/eda6c63111fd4017823f8ebcdd16b6ed : SUCCESS in 46m 16s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/8fa54b9068414e3087fb4ace19bf2f8c : SUCCESS in 46m 21s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/d7c40067e0394cf0a7a067e4891f9501 : SUCCESS in 35m 51s\n- rally-task-murano https://zuul.opendev.org/t/openstack/build/0ea8e11d5805435bb92e296cd61a0ebe : FAILURE in 40m 09s (non-voting)\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/c2813c0c42c84d288b341b3efeeb3a37 : SUCCESS in 50m 37s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/34487f63ee84488e8348c0a2b387dfb8 : SUCCESS in 41m 58s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/c8cfe36f92494c83b8a2469c7230b81d : FAILURE in 1h 01m 53s (non-voting)\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/6964cece7f674a96bce242cfcbed05c1 : SUCCESS in 44m 58s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/849c08405e7d483f945ded762d1de8ad : SUCCESS in 39m 45s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/fe1ba2d41c354b6aa722d1acb9bc1622 : SUCCESS in 41m 18s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/9b8db5baedab45e28dc8ab169d345015 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":1},{"id":"20fe59434a02d25167b08fb7476f8a8845a58ce1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-10-12 13:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecf5c59f91ae83b48dbce8c35d7bd247c7f002cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 13:10:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e30d678e2b7c3c58f0eaa6b3d8d9b7fe51c3131","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 14:45:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"554633cc1ec1271751b206688b42a8d52ac617e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 14:45:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/5822794bf6da4a5583248012ba0ed729 : SUCCESS in 5m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8b662f3fc94ad89f36ed8a892802d7 : SUCCESS in 5m 23s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/7b2783db1e5648c590a6f69585246ed6 : SUCCESS in 4m 54s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/83348349c7cd4c66b61a8dec22577daa : SUCCESS in 5m 25s\n- rally-tox-py38 https://zuul.opendev.org/t/openstack/build/f79f01a706104b98ba38f37c6127b422 : SUCCESS in 4m 43s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/360c17d00bb4422cbf202018c42a6b1c : SUCCESS in 45m 20s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/6a8e618cf1a24fa2a0bf97c274ec3082 : SUCCESS in 9m 15s\n- rally-task-basic-with-existing-users https://zuul.opendev.org/t/openstack/build/1be307ffb3aa45f597ba91ded30bc7cd : SUCCESS in 40m 00s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/957b911a9f564cc5a5d519c3490f5990 : SUCCESS in 32m 36s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/df9819473ae14e61a55acce35035037a : SUCCESS in 1h 01m 04s\n- rally-task-ironic https://zuul.opendev.org/t/openstack/build/bf228df56352445585873ad6e657eddf : SUCCESS in 50m 40s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/e8edfb704084436f9d347531b8b318e6 : SUCCESS in 48m 37s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/278eee2c276b4a94a022980d2428ccd1 : SUCCESS in 41m 25s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/b65c673ad973449290cae46b0bacaae8 : SUCCESS in 1h 27m 40s\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/6a557d1dbbe0423d80c27fda936ff2eb : SUCCESS in 35m 28s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/420292739077449da4952ff24c945c3c : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"608163f52552ff76b316f9b62a903c947d6fee20","revisions":{"608163f52552ff76b316f9b62a903c947d6fee20":{"kind":"REWORK","_number":1,"created":"2020-10-12 11:50:36.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/45/757545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/45/757545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/45/757545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/45/757545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/45/757545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/45/757545/1"}}},"commit":{"parents":[{"commit":"91335b60a021d79e8256353a2e99a3ef8d6f27bd","subject":"Fix an erroneous grammar in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/91335b60a021d79e8256353a2e99a3ef8d6f27bd"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-10-12 11:50:13.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-10-12 11:50:19.000000000","tz":180},"subject":"Fix used python versions for CI","message":"Fix used python versions for CI\n\nChange-Id: Ib3ce073e260c1387f7d26b7d994da2fdc8077e06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/608163f52552ff76b316f9b62a903c947d6fee20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/608163f52552ff76b316f9b62a903c947d6fee20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
