)]}'
{"id":"openstack%2Frally-openstack~994259","triplet_id":"openstack%2Frally-openstack~master~I227df37aabf27fd2123028ec96fbec13f92fd283","project":"openstack/rally-openstack","branch":"master","attention_set":{},"removed_from_attention_set":{"39237":{"account":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"last_update":"2026-07-24 13:00:26.000000000","reason":"Change was submitted"},"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-07-24 11:00:13.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-07-24 13:00:26.000000000","reason":"Change was submitted"},"36989":{"account":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"},"last_update":"2026-07-24 10:18:55.000000000","reason":"\u003cGERRIT_ACCOUNT_36989\u003e replied on the change","reason_account":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}}},"hashtags":[],"change_id":"I227df37aabf27fd2123028ec96fbec13f92fd283","subject":"Fix load balancer deletion race condition during create/update task","status":"MERGED","created":"2026-06-22 11:35:48.000000000","updated":"2026-07-24 13:00:26.000000000","submitted":"2026-07-24 13:00:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"994259","meta_rev_id":"d8927030ee8d19391a70623c6c7e00071f8b0eaf","_number":994259,"virtual_id_number":994259,"owner":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-24 13:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-07-24 11:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-07-24 10:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-07-24 11:28:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 12:13:19.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-06-22 13:02:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-23 06:27:16.000000000","updated_by":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"},"reviewer":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"},"state":"REVIEWER"},{"updated":"2026-07-01 13:49:01.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"CC"},{"updated":"2026-07-24 11:00:13.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"96dbc1d7058ab23a60e2dfef0be2b948d828ea7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"date":"2026-06-22 11:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d96cff7c09bde69b2ab048b8be65ea3639a2ae60","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-06-22 12:13:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8736e5c4b572f55ee1f3cb4d6f59e743095f594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 13:02:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae2c0cd906f4acd9b0860be98701281\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e0a52cd5d7a14d458ead82665a770524 : SUCCESS in 5m 40s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/4c81acd6d5cd40c4b927b9972e033b2a : SUCCESS in 3m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a53903c9804f83a41c217979861f18 : SUCCESS in 2m 26s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd203a835e94a7da0ee66aaeec09cb0 : SUCCESS in 2m 21s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/0a5de931849b4ad38762b37d9b5183a3 : SUCCESS in 2m 42s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/e70b46769ba64e58b55be00a7f4372b5 : SUCCESS in 3m 50s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/04cd906403c943878e1f7959aaea3441 : SUCCESS in 3m 44s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/0165fa31960c4ffe8d251d95da065441 : SUCCESS in 34m 32s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/629a29ad27f242daa23b4f62b04d71f6 : SUCCESS in 8m 13s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/7d394a5076554c58ba028d9522a9352e : SUCCESS in 30m 25s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/71cf9e9fdfbd4c4a93b2bc4c039101d7 : SUCCESS in 1h 05m 50s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/90683a7180d0404497f200af33703be8 : SUCCESS in 43m 30s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/9cdb47a194a5434898f910680e41e1f2 : SUCCESS in 33m 35s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/2fe4f31327494a98a9836a1ee37cea54 : SUCCESS in 34m 08s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/0d46309fe3d147fa9612c656a3064030 : SUCCESS in 29m 50s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/d9a2cdf8bff0467fae9f1a14714011b7 : SUCCESS in 37m 20s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/fdede207f9394937a9b7eb211c9bb629 : SUCCESS in 1h 20m 37s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/43148da9f9294e55b6f609a486de62d5 : SUCCESS in 31m 09s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/02aa13a4e5c44441be29b1ea994aef66 : FAILURE in 33m 31s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/08a097317a414f6bbbc7314bfd25541a : SUCCESS in 30m 16s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/8ee1db892f63402ca7e9819d40b58638 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"25d070487ccfacc9730143f0accb9a499b46b550","author":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"},"date":"2026-06-23 06:27:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"671d302d6bae6515f64e0d2cfd4dd9b5348a3a78","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-01 13:49:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8d17688967b38332e00085d653863db81dd76be5","author":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"date":"2026-07-24 09:56:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00ec6a6b25026a8d69cbab349dc841d112f90622","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"date":"2026-07-24 09:57:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"93e9931802e3aa4aaef05e7be39c3c1573bed296","author":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"},"date":"2026-07-24 10:18:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1254c299d7f59d2c8ad655ae505aea955a55984","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-24 11:00:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dc8e3664e0f8055108e1c6cd8bff73256298ea4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 11:12:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9c6685a8a54df5a630765c59b1e6c4\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bddce1c513f14575ae758649687e086c : SUCCESS in 5m 55s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/18777390072b46baa1cf17a2e1183c04 : SUCCESS in 3m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c52a3eecf54bb1a33ccb457e0683a7 : SUCCESS in 2m 17s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/4db7079e871741f7b80daa252aa2309b : SUCCESS in 4m 02s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/fae37ef01a44418e8d60d8e82e4d9b62 : SUCCESS in 3m 36s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/072223f10165470f9b9c712ef87cfc20 : SUCCESS in 4m 13s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/43b5b8a05f5b455198f3b3a5bd624676 : SUCCESS in 3m 42s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/563d3b76de9345c391eda2e7a6c14462 : SUCCESS in 45m 13s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/3dc0dcf00e024194ab88d6fcfdb67745 : SUCCESS in 8m 32s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/1da2daa6f75846cc8a4cf7313132391b : SUCCESS in 37m 01s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/a899f46e89bf460c90545ca1c6c8f7cc : SUCCESS in 1h 08m 42s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/57d60bfce65d456da87b75599926cc6c : SUCCESS in 42m 53s\n- rally-task-manila-no-ss https://zuul.opendev.org/t/openstack/build/1f55b3c7a6444c39a46631254d04f08d : SUCCESS in 35m 06s (non-voting)\n- rally-task-manila-ss https://zuul.opendev.org/t/openstack/build/afce28ccba6c42e896d9fb332ea830fe : SUCCESS in 37m 53s (non-voting)\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/c4bd100ba2074bc792215a6cf4e5b4c3 : SUCCESS in 34m 34s\n- rally-task-neutron-with-extensions https://zuul.opendev.org/t/openstack/build/60ded64af5c64ab6aa3387e9640df088 : SUCCESS in 36m 20s (non-voting)\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/687242979a0d4c809bdbbb0b8ebdcca2 : SUCCESS in 1h 06m 46s\n- rally-task-telemetry https://zuul.opendev.org/t/openstack/build/9fa6ba77540c4d0cab7414dc5db5416b : SUCCESS in 35m 50s (non-voting)\n- rally-task-watcher https://zuul.opendev.org/t/openstack/build/67ee2100d1664bcc9fd38b197e932463 : SUCCESS in 32m 42s (non-voting)\n- rally-task-zaqar https://zuul.opendev.org/t/openstack/build/577d78251bd645dd96a786cda526816a : SUCCESS in 30m 48s (non-voting)\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/dd6d62adf1a148eb9cae410f31a6c4b3 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":2},{"id":"184fb32d0ed70660875394b0753d0d599b9000e8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-07-24 11:28:39.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56e514ffadf0c16d33ed95ac0644d22d0b23734d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 11:29:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f55ddfbadb51eace800a6d52b39415f477b530c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 13:00:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b8e78f2aaf740849ef2d9210a9dfe8b\n\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c170c653ed514b12beb9b809a0e1867e : SUCCESS in 6m 00s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/c6d6b4748dc64408b0e80b126fab98c1 : SUCCESS in 4m 25s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6877ad2c7b48dead31b45ae0ac7975 : SUCCESS in 2m 10s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/96e6e76cf58c4ec5bc46716e0b419dd2 : SUCCESS in 3m 30s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/ec55483f4c434f74af3a522a5d63e1ce : SUCCESS in 3m 49s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c81dc6e2e3614181aadb107cc3fb3be7 : SUCCESS in 4m 28s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/0f2471ed6dd1458889ba03bfcc6014b2 : SUCCESS in 5m 32s\n- rally-dsvm-tox-functional https://zuul.opendev.org/t/openstack/build/4d1da9f2fae2496d838c00a0293e39c2 : SUCCESS in 41m 37s\n- rally-openstack-docker-build https://zuul.opendev.org/t/openstack/build/459fc7ab47d74668939b5e75bf2736c0 : SUCCESS in 8m 27s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/302f9fdee9a54e43815076556847a831 : SUCCESS in 33m 24s\n- rally-task-cinder https://zuul.opendev.org/t/openstack/build/b073da4c63384c4bbedb6fe5baf12884 : SUCCESS in 1h 18m 01s\n- rally-task-keystone-glance-swift https://zuul.opendev.org/t/openstack/build/09dc73066df64dd2a89b902c6d733137 : SUCCESS in 45m 36s\n- rally-task-mistral https://zuul.opendev.org/t/openstack/build/d8b8351f4c99487f8b8c222b6ae7109d : SUCCESS in 16m 40s\n- rally-task-neutron https://zuul.opendev.org/t/openstack/build/1a3e12626bae4b01b490ca4f30bf00eb : SUCCESS in 32m 24s\n- rally-task-nova https://zuul.opendev.org/t/openstack/build/8aa888539a9e49aeaecc13d9e5642edf : SUCCESS in 1h 08m 09s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/8c4459105d374f0ba56b6b9e638befb8 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"d8927030ee8d19391a70623c6c7e00071f8b0eaf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 13:00:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8bca1b9aca23d4e25819cd63b8538b2c92fb141","revisions":{"48019a2f99fdee23424bd19f415f22e21162c9ff":{"kind":"REWORK","_number":1,"created":"2026-06-22 11:35:48.000000000","uploader":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"ref":"refs/changes/59/994259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/59/994259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/1"}}},"commit":{"parents":[{"commit":"97e84637b02bf991f30091d8a35947b3aaf516a1","subject":"Fix flaky test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/97e84637b02bf991f30091d8a35947b3aaf516a1"}]}],"author":{"name":"Alaa Aldandachi","email":"alaa.dan.85@gmail.com","date":"2026-06-22 11:18:53.000000000","tz":120},"committer":{"name":"Alaa Aldandachi","email":"alaa.dan.85@gmail.com","date":"2026-06-22 11:31:11.000000000","tz":120},"subject":"Fix load balancer deletion race condition during create/update task","message":"Fix load balancer deletion race condition during create/update task\n\nThe create_and_update scenario triggers an update for the LB name, but\nit doesn\u0027t wait for ACTIVE status. This causes cleanup manager to immediately\nattempt to delete the LB, resulting in HTTP 409 error.\n\nChange-Id: I227df37aabf27fd2123028ec96fbec13f92fd283\nSigned-off-by: Alaa Aldandachi \u003calaa.dan.85@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/48019a2f99fdee23424bd19f415f22e21162c9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/48019a2f99fdee23424bd19f415f22e21162c9ff"}]},"branch":"refs/heads/master"},"a8bca1b9aca23d4e25819cd63b8538b2c92fb141":{"kind":"REWORK","_number":2,"created":"2026-07-24 09:57:42.000000000","uploader":{"_account_id":39237,"name":"Alaa Aldandachi","display_name":"Alaa","email":"alaa.dan.85@gmail.com","username":"aa2165"},"ref":"refs/changes/59/994259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally-openstack","ref":"refs/changes/59/994259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally-openstack refs/changes/59/994259/2"}}},"commit":{"parents":[{"commit":"97e84637b02bf991f30091d8a35947b3aaf516a1","subject":"Fix flaky test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/97e84637b02bf991f30091d8a35947b3aaf516a1"}]}],"author":{"name":"Alaa Aldandachi","email":"alaa.dan.85@gmail.com","date":"2026-06-22 11:18:53.000000000","tz":120},"committer":{"name":"Alaa Aldandachi","email":"alaa.dan.85@gmail.com","date":"2026-07-24 09:57:24.000000000","tz":120},"subject":"Fix load balancer deletion race condition during create/update task","message":"Fix load balancer deletion race condition during create/update task\n\nThe create_and_update scenario triggers an update for the LB name, but\nit doesn\u0027t wait for ACTIVE status. This causes cleanup manager to immediately\nattempt to delete the LB, resulting in HTTP 409 error.\n\nChange-Id: I227df37aabf27fd2123028ec96fbec13f92fd283\nSigned-off-by: Alaa Aldandachi \u003calaa.dan.85@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/a8bca1b9aca23d4e25819cd63b8538b2c92fb141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally-openstack/commit/a8bca1b9aca23d4e25819cd63b8538b2c92fb141"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36989,"name":"Christian Mattsson","email":"christian.mattsson@cleura.com","username":"jeeaaasus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
