)]}'
{"id":"openstack%2Frally~1000190","triplet_id":"openstack%2Frally~master~I6c1e5d542337c1f9c2a4586fa8ef2c927b94cec3","project":"openstack/rally","branch":"master","topic":"revert","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-08-08 18:51:16.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"I6c1e5d542337c1f9c2a4586fa8ef2c927b94cec3","subject":"Revert \"Remove deprecated modules\"","status":"MERGED","created":"2026-08-08 10:44:18.000000000","updated":"2026-08-08 21:17:57.000000000","submitted":"2026-08-08 21:17:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000190-revert","meta_rev_id":"3d2917a0f61e8efee092fca186e40e48cb76c9e7","_number":1000190,"virtual_id_number":1000190,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-08 21:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-08-08 18:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-08-08 18:51:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-08 11:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e245f9fdb322dd2d43545cce0775eff606fb286f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 10:44:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bec9b17c53ec9fdf6974891e3c12821ec766c53","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 10:44:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ffff08e64605c204dbf8663c245d23c1688c28e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 11:15:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93ef23e95e7f42128aa9040e213cca34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45650cc91c6244d793a10bae92690622 : SUCCESS in 4m 40s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ddaf4dae39f7432a9a830c7c813de2f6 : FAILURE in 4m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/2318892e0e4d47b88619ace9bb32cc0e : SUCCESS in 3m 45s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/be2b4609b6524d408930547e7c7bdb80 : SUCCESS in 2m 02s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1899e72f5bc64f5e982df0ce5cf8e7c4 : SUCCESS in 2m 19s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b439da7c26374882ae5e3633042c9e50 : SUCCESS in 3m 06s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/b5cfbc6d59484234b8a4d8be9e7f131b : SUCCESS in 2m 51s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/09e5a420bb9a4d9fad489d1c5c312298 : SUCCESS in 1m 54s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/339a060bf9af4976869e96eef9fb5b9c : SUCCESS in 2m 54s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e600dbe212b540219dbedd5e2e2bc47c : SUCCESS in 2m 35s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2581ff606a8942a5be1bc2e77f3e875f : SUCCESS in 6m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/275520b25e014461b8b4545051d58ab8 : SUCCESS in 6m 17s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88a3020a601a48a69dd4e20055a8aba3 : SUCCESS in 7m 45s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d5dc06feb8e4ae4bd8f937429098a88 : SUCCESS in 5m 07s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/95783d65b737451ea4a2e4752e7b4b60 : SUCCESS in 6m 24s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/1eddd24208b247fe81671418ef592afb : FAILURE in 25m 37s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/eec59bc39d2748429fff86a696fc4188 : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":1},{"id":"60f860fa92e87cea694f380aba8413e90e0ab8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 12:20:00.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"40896bb62ee49365bfeca645d92a9b5e5b9edb64","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 12:20:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"011e4d301b6d91863d0d929f6a15ac459b0bc120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 12:57:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c7b57d302984f3f85c5d987c00120d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a45e9d3c3b74e99ac99599458e91997 : SUCCESS in 4m 02s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e44280f794064aafb8040a067153a00c : SUCCESS in 2m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f2f0af719e7d41b6b3beee75735e6eec : SUCCESS in 3m 39s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/3f3e92290f044dc8a7ca92dd75427c95 : SUCCESS in 3m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1850e0fb3f6f4d51aad0468ee8d5d1f3 : SUCCESS in 1m 33s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/08c9ac37454d445499b04127d2bfba8c : SUCCESS in 2m 48s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/18d0f8acb6a74512aa11e9bc784a714d : SUCCESS in 2m 45s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/1620e99d12684bb4be89a4dc25096b82 : SUCCESS in 1m 55s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/6f07b21cd64547e38b682009acddd017 : SUCCESS in 3m 00s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/24bf36da24d741378ac2a319d83f077e : SUCCESS in 2m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/511864a14b294d4381afb220186d7852 : SUCCESS in 4m 25s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/225b8873ab1d4327b9a1fd9c6e0c55de : SUCCESS in 5m 32s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90aa390550c44d6897ca533d358af589 : SUCCESS in 5m 59s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aed0768d42c49fd9de258b8b3ca0fe2 : SUCCESS in 5m 00s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/17f9387ea6a844d4aa072f7f36a1c5a0 : SUCCESS in 9m 20s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/2bde02afee694f4c970b0049bf0e2bf6 : FAILURE in 27m 08s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/c743bf8bfb664120a3d5e50bb1f77fae : SUCCESS in 14m 48s","accounts_in_message":[],"_revision_number":2},{"id":"7d4549cb57773358d2f90b143d8ea3b77789e33d","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 15:44:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed6d70fe4646d543d8081d1d5e96b8a870f1421b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 16:24:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f480366e4957487da4244ada19752c0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c441838586449f898e513ff5666a781 : SUCCESS in 4m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/eb4a89123fc8431b80483e0dfcd283a5 : SUCCESS in 4m 35s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/87901a58b71c487d8a1937b2484b2a12 : SUCCESS in 3m 40s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/734c56ccc4004967b989c1ea0ed6e6c3 : SUCCESS in 5m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcaba28fdc6c4bdda51e6278f33869f8 : SUCCESS in 3m 03s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/b09948eb34364963b4473a329aa4cc87 : SUCCESS in 1m 52s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/78e4b6622a684d8e89692d3eebda8690 : SUCCESS in 2m 52s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/9675387fd84143908f2512a89932164d : SUCCESS in 3m 09s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/47ebfdbd8d75495782ac1c3f652c04f5 : SUCCESS in 2m 38s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/32bda713564c46c68f971f9845b03ae0 : SUCCESS in 2m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7a9548b1359d486591304af2fa3483f4 : SUCCESS in 6m 37s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/f153ba5a45184f4f8a18e02a1009c026 : SUCCESS in 6m 20s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bd4bd1d4dbb44b1ab524931bb0e416f : SUCCESS in 6m 06s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/705ccfabd1784560a2dfb51eb160757b : SUCCESS in 6m 03s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/228db7e544d84cd09ee1d54530607ced : SUCCESS in 8m 01s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/d528660d1ab5481599d32bdd2d28903f : FAILURE in 16m 39s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/5c4ae19bc5ee4564a8f556b78c8a631f : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":2},{"id":"93ff42666b26ae5607e846b25658939347334a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 18:51:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e00cf8197c8b3dc721d05f938b28037301eb6ab2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-08 18:51:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d22b8dad66c9180885d81ed303e22d9fbae6c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 19:27:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600dad88a5264774b6ebe262ca625624\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7bf6726c6f4bcbb130dfd54dc17fae : SUCCESS in 2m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f033ea77a6b14dd6aa202cf14227ebc3 : SUCCESS in 5m 03s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b2953df5c7b7418db6faf4ee3f5b9779 : SUCCESS in 3m 52s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/75f4a11d679b4de19e4c9c12158075e3 : SUCCESS in 2m 41s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e67137adde478d9cc52ffc5a7a6c4c : SUCCESS in 2m 37s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/047817f44b91498fa8e83ae4131e8fff : SUCCESS in 3m 22s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/0ca6d075983b4df5a09373bdc65a6fbb : SUCCESS in 1m 58s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/fadbcdaa72084e7d9ec80a1f7d4e8fcc : SUCCESS in 3m 04s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/dc0221f8b3314e8da8d999ada4b9c95a : SUCCESS in 1m 49s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/df3da4224fb94a6c9c39a0d46791da65 : SUCCESS in 2m 31s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d44a9ccc704f4c42a8a7d0e7dd1ab517 : SUCCESS in 4m 35s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2047db84e1ea45f2aab8c142655eec81 : SUCCESS in 5m 57s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67daddba028f45e59dd19073e1fa1c2b : SUCCESS in 6m 25s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6cb723155c74774bd460e58880d6135 : SUCCESS in 5m 56s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/89cfb83be0ef4465b1f1a70976c581eb : SUCCESS in 7m 42s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/16e70c1abb114f03bf284c2b6e6585a8 : SUCCESS in 17m 18s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/ad3c3846aac0436e890244a24bd21d6f : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4d25858a631b700dccb6db6bee933f22e3003618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 20:33:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8cd01684147368dad59be851a529d3dbc993107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 21:17:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4d6a37f625048e3831725d4bdb95881\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd3b160147d440f1bee9b29d12e7047a : SUCCESS in 4m 24s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a7d98d5ae5a0457bb3f3657144c4df98 : SUCCESS in 2m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9d5f60b2774848b9a3d7b492f1da9fee : SUCCESS in 3m 33s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/bddb163eb25a4739a5b1020384138d47 : SUCCESS in 2m 53s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d34e42c31c54c85a5b4aef2e74e7467 : SUCCESS in 2m 24s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/624cc185c012460ab4b9fe614c6a8942 : SUCCESS in 2m 56s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/1c8e44dd4b764998a1ad147759170fd6 : SUCCESS in 2m 43s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/c6336cf8fe964b458576ac9b1452feee : SUCCESS in 2m 47s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/218ee10030bf4fd7a83dc27ae806a174 : SUCCESS in 3m 20s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ab1b82ffc05f49a780b65c5bbd23d9be : SUCCESS in 3m 11s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/73c8af17d3df4f67aec8d003b94ac30b : SUCCESS in 5m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1ee346bb210646b7bbc768963311f56c : SUCCESS in 6m 29s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e434274fba3747c6b147ff061f8909d1 : SUCCESS in 5m 31s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d363648d78234bc7891da05e65e14a92 : SUCCESS in 6m 06s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/05a5d50ef39143699fc3eb312496cbb9 : SUCCESS in 8m 17s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/ba7a33b3c9d94b35895388d9eaf5cb54 : SUCCESS in 28m 19s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/0bc231196ae3411583604784cbd0a46d : SUCCESS in 33m 25s","accounts_in_message":[],"_revision_number":3},{"id":"3d2917a0f61e8efee092fca186e40e48cb76c9e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 21:17:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2383d7680d45baf8bfa0c02734a4f5ad444a4f63","revisions":{"41b8f2f9e2cec511ebae71ae2cfe16be33db2235":{"kind":"REWORK","_number":1,"created":"2026-08-08 10:44:18.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/1000190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/1000190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/1000190/1"}}},"commit":{"parents":[{"commit":"f794a0d0594099d0adc4f63299b95f4f1cd5d4a8","subject":"Merge \"[rtd] Fix build from latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f794a0d0594099d0adc4f63299b95f4f1cd5d4a8"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 10:36:33.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 10:44:05.000000000","tz":120},"subject":"Revert \"Remove deprecated modules\"","message":"Revert \"Remove deprecated modules\"\n\nThis reverts commit 79787209e1232e6bc78a9fd5f32012626927ed4a.\nThe original commit is not wrong, we just want to post-pone backward\nincompatible changes to the next release\n\nChange-Id: I6c1e5d542337c1f9c2a4586fa8ef2c927b94cec3\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41b8f2f9e2cec511ebae71ae2cfe16be33db2235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41b8f2f9e2cec511ebae71ae2cfe16be33db2235"}]},"branch":"refs/heads/master"},"ea828bdbd5d7da0869c56b51816c1ef1a96923a3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-08 12:20:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/1000190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/1000190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/1000190/2"}}},"commit":{"parents":[{"commit":"f794a0d0594099d0adc4f63299b95f4f1cd5d4a8","subject":"Merge \"[rtd] Fix build from latest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f794a0d0594099d0adc4f63299b95f4f1cd5d4a8"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 10:36:33.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 12:20:00.000000000","tz":0},"subject":"Revert \"Remove deprecated modules\"","message":"Revert \"Remove deprecated modules\"\n\nThis reverts commit 79787209e1232e6bc78a9fd5f32012626927ed4a.\nThe original commit is not wrong, we just want to post-pone backward\nincompatible changes to the next release\n\nPRAGMA: NO COVER\n\nChange-Id: I6c1e5d542337c1f9c2a4586fa8ef2c927b94cec3\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ea828bdbd5d7da0869c56b51816c1ef1a96923a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ea828bdbd5d7da0869c56b51816c1ef1a96923a3"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2383d7680d45baf8bfa0c02734a4f5ad444a4f63":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-08 18:51:11.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/90/1000190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/90/1000190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/90/1000190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/90/1000190/3"}}},"commit":{"parents":[{"commit":"ba8aabf19073736b4664c8d02fca3f4b313ba1e7","subject":"[ci] Fix rally-task-simple job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ba8aabf19073736b4664c8d02fca3f4b313ba1e7"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 10:36:33.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-08 18:51:11.000000000","tz":0},"subject":"Revert \"Remove deprecated modules\"","message":"Revert \"Remove deprecated modules\"\n\nThis reverts commit 79787209e1232e6bc78a9fd5f32012626927ed4a.\nThe original commit is not wrong, we just want to post-pone backward\nincompatible changes to the next release\n\nPRAGMA: NO COVER\n\nChange-Id: I6c1e5d542337c1f9c2a4586fa8ef2c927b94cec3\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2383d7680d45baf8bfa0c02734a4f5ad444a4f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/2383d7680d45baf8bfa0c02734a4f5ad444a4f63"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"f794a0d0594099d0adc4f63299b95f4f1cd5d4a8","ours":"ea828bdbd5d7da0869c56b51816c1ef1a96923a3","theirs":"ba8aabf19073736b4664c8d02fca3f4b313ba1e7","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
