)]}'
{"id":"openstack%2Frally~1000447","triplet_id":"openstack%2Frally~master~I56d80d114b31615ca2eeca8879c0245bcea3e840","project":"openstack/rally","branch":"master","topic":"prepare-5.1.0","hashtags":[],"change_id":"I56d80d114b31615ca2eeca8879c0245bcea3e840","subject":"Propose Rally 5.1.0 release","status":"MERGED","created":"2026-08-11 12:26:49.000000000","updated":"2026-08-11 12:52:10.000000000","submitted":"2026-08-11 12:52:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000447-prepare-5.1.0","meta_rev_id":"3887eec67d2e6c8f1cdbbe901632ea84104d6df3","_number":1000447,"virtual_id_number":1000447,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-11 12:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-08-11 12:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-08-11 12:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-11 12:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b7b089bd062c42bf13e5a890b488dab503745cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-11 12:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b7f834c0ca2d0943e0e998258544e41eccf9e10","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-11 12:27:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa0588966b0a02469d16daf562c3b02668e229ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:40:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f07f1187a94488ad92a2e949e777e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49adc57b809c458e9b37b990768c3859 : SUCCESS in 6m 03s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/1831477b9bfc44e08603536b5069d3a0 : SUCCESS in 5m 24s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9737aef59d8447a3a33b13f1cbbb3b35 : SUCCESS in 4m 00s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/48a5b2e6187d48ae98109b3646372a57 : SUCCESS in 3m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f078e3148d4e9eb07b4d84a4ed60f7 : SUCCESS in 2m 43s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a9fef72bf0cf4e57950dc1ff51a050f4 : SUCCESS in 3m 37s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/713b0448042d49279bd90ab729765106 : SUCCESS in 2m 55s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/86b8b636aa464c12ba718e8af0d8947b : SUCCESS in 4m 09s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/2557857608654256be00d9976b2ff708 : SUCCESS in 4m 44s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/f1559ca1ed8845b399a1a1fd35f4036b : SUCCESS in 3m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ef7ed8b8c8ce4021ac885d889b30312e : SUCCESS in 7m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7d3fdac7d59048218c5648092ad2bbb1 : SUCCESS in 6m 32s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fd004e325ca4448bc8bf7385fa8d3bb : SUCCESS in 5m 10s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eb04d579bde4a1baff947c2c2ed95f8 : SUCCESS in 6m 24s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/d8c4f067f833434e8216847b1de391b1 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"93f22fc08a44e6221a635e1af835ad972a75c67f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:40:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0f8b5b8a9bcb263f3fa06fbbfae38070dcdf398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:52:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabd41bfd0c548c8a26c9ec1f1efd9bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2897d70ed6f0481f8cdbe2f137ce8879 : SUCCESS in 3m 29s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/46897416217946f1bcb5ef45abd9b62f : SUCCESS in 5m 37s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/24424afe7f46449b8fe840386bcff91d : SUCCESS in 3m 53s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/6475db78c26942d5a3e2a89a345b09ca : SUCCESS in 1m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e84e5e7bd94e0db3cda0c52fb86ebd : SUCCESS in 1m 57s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/8b2c585ad4c643dead766f3e1091a7db : SUCCESS in 3m 40s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/20aee3ef3be248efbfcaac86e95541dd : SUCCESS in 3m 04s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ae0148245335453bab686aa1ef1f51b2 : SUCCESS in 3m 02s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/8c0b910a5caa46fba07081194550c4d6 : SUCCESS in 2m 57s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ac252a7e72774f179617c59a53d3a7fd : SUCCESS in 2m 33s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/725c5eb2f252463ea3527a4d02c7a2a3 : SUCCESS in 6m 38s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/59a5a273607a43b699596d604dcd81d0 : SUCCESS in 6m 30s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b775e9bb93f04caf819da7ee4ad1b487 : SUCCESS in 5m 42s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70e2e30e069c4c65b9ef2bcfd3d8b5f7 : SUCCESS in 6m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2b2562ab8a414a69a9420597d90b5ba0 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3887eec67d2e6c8f1cdbbe901632ea84104d6df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 12:52:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6d1e2b224d464a8e34fb31d977841ec6c0535e9","revisions":{"f6d1e2b224d464a8e34fb31d977841ec6c0535e9":{"kind":"REWORK","_number":1,"created":"2026-08-11 12:26:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/47/1000447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/47/1000447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/1000447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/1000447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/1000447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/47/1000447/1"}}},"commit":{"parents":[{"commit":"1fa5a52f6172685a6a2b6bd60df8b7294c15ab11","subject":"A couple docs improvements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1fa5a52f6172685a6a2b6bd60df8b7294c15ab11"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-11 12:26:34.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-11 12:26:42.000000000","tz":120},"subject":"Propose Rally 5.1.0 release","message":"Propose Rally 5.1.0 release\n\nChange-Id: I56d80d114b31615ca2eeca8879c0245bcea3e840\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f6d1e2b224d464a8e34fb31d977841ec6c0535e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f6d1e2b224d464a8e34fb31d977841ec6c0535e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
