)]}'
{"id":"openstack%2Frally~1000702","triplet_id":"openstack%2Frally~master~Ie1b7af64921bf23c5a2c544ef3aa3cb8cf1a2f04","project":"openstack/rally","branch":"master","topic":"fix_plugin_doc","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-12 11:06:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie1b7af64921bf23c5a2c544ef3aa3cb8cf1a2f04","subject":"Fix outdated import paths in plugin documentation","status":"MERGED","created":"2026-08-12 09:13:36.000000000","updated":"2026-08-12 11:06:29.000000000","submitted":"2026-08-12 11:06:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000702-fix_plugin_doc","meta_rev_id":"9cd073ffbb8cab18dbea730ce864780f73630bc5","_number":1000702,"virtual_id_number":1000702,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-12 11:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-08-12 10:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-08-12 10:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-12 09:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-12 10:53:52.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"48e643636ce8bb5cf8bc00f50fda54b4962341f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-12 09:13:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4006992bac68cd02827f0156023a8bf73e8f89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 09:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/252d3fbd22bc4b6883fbe892cee78955\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce36468212c4397be4598c046b42607 : SUCCESS in 3m 04s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6015b050e0e147a2a5d26f3d9d61fb2f : SUCCESS in 4m 17s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/9eb6aaf59a46468490e2949f951418dc : SUCCESS in 3m 46s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/0e54a83ac8bf42d69cc2d7452094fd29 : SUCCESS in 3m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/569e943ef9234c39a6bd17a18e7afe55 : SUCCESS in 3m 45s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/ce2cbe35ed4a40808a4ac994cc993aed : SUCCESS in 2m 36s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/f0641000ca6144899a329806086dfa5b : SUCCESS in 2m 57s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/af171f47deb1418fa12e435faac050b8 : SUCCESS in 3m 54s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/ccf8192d430846ef9d79c240cd06fc48 : SUCCESS in 3m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/ed049cea92634990a1d082e9860ee008 : SUCCESS in 2m 42s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/97612d99e7214eceae1a178483cf7950 : SUCCESS in 6m 14s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e13437879b7d40d4ab245a3d32179c82 : SUCCESS in 5m 07s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06809eb4eadb4b2fb7d9ec4823ddb7e0 : SUCCESS in 5m 27s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53fe045b2e4f450ea342114689af0bb9 : SUCCESS in 6m 09s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/125bbb0708894c7a9fc4c31df2006cff : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"886e7786e221c9480fc975333ff0df8ad02f93f7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-12 10:53:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"077075cc062b90f90865570901057c3946c7135d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 10:54:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"034f48354aac0cccb3f1ea7e825abf6e858452ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 11:06:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bc94f1c7ccb4c66aa9880bf222cde1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c8b12ba8d644b89deb3178cc736522 : SUCCESS in 4m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/332ff56e99c64fbabfc458415190eb08 : SUCCESS in 4m 26s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/abeb0c920935430bbd88d7c5d5f41e96 : SUCCESS in 4m 00s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/c24a0312b7514f9c83eacf0ca6add640 : SUCCESS in 3m 09s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/31fdc644f8d349a4adfd3fd6d74e4f92 : SUCCESS in 1m 56s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/f7cf68758f484d2a85da8633db3a3ac4 : SUCCESS in 3m 50s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/77cddd3873d34955b88abb5aa4b33b0f : SUCCESS in 3m 22s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/6f2354115fda49deb7c2221525e924c3 : SUCCESS in 3m 05s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9729b8607a9741a2ada033e91dc630fd : SUCCESS in 3m 00s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/25a4695d4a73415d90fa10221e07ab6b : SUCCESS in 2m 41s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/2caa1807ac6d40c1af4bc7c826b4e31e : SUCCESS in 6m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1f9f67ad21c144278d269595e71a7d36 : SUCCESS in 5m 21s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cff44506e988405c9f6ea9a48ff958b4 : SUCCESS in 5m 25s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe60faa788ae496cbbfd2a56527c2280 : SUCCESS in 5m 54s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c88091990f2946f28fafe50a0267ef9c : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9cd073ffbb8cab18dbea730ce864780f73630bc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 11:06:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69fecc94453495406c13bfa1ea888c301387f35a","revisions":{"69fecc94453495406c13bfa1ea888c301387f35a":{"kind":"REWORK","_number":1,"created":"2026-08-12 09:13:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/02/1000702/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/02/1000702/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/1000702/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/1000702/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/02/1000702/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/02/1000702/1"}}},"commit":{"parents":[{"commit":"f1495df14fbe119d370b56a867415b696e32711c","subject":"Correct changelogitem for rally-openstack 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f1495df14fbe119d370b56a867415b696e32711c"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-12 09:07:53.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-08-12 09:09:40.000000000","tz":120},"subject":"Fix outdated import paths in plugin documentation","message":"Fix outdated import paths in plugin documentation\n\nThe scenario and context plugin docs still referenced\nrally.plugins.openstack which was moved to the separate\nrally-openstack package. Update imports to use the correct\nrally_openstack.task and rally_openstack.common paths.\n\nAssisted-By: claude-opus-4.8\nRelated-Bug: #2163189\nChange-Id: Ie1b7af64921bf23c5a2c544ef3aa3cb8cf1a2f04\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/69fecc94453495406c13bfa1ea888c301387f35a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/69fecc94453495406c13bfa1ea888c301387f35a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
