)]}'
{"id":"openstack%2Frally~1000720","triplet_id":"openstack%2Frally~master~I524e77bc6ce13825421b3bb7820a743b0dc8ecaf","project":"openstack/rally","branch":"master","topic":"fix-doc-publishing","hashtags":[],"change_id":"I524e77bc6ce13825421b3bb7820a743b0dc8ecaf","subject":"[docs] Fix publishing triggers","status":"MERGED","created":"2026-08-12 13:00:53.000000000","updated":"2026-08-12 14:34:53.000000000","submitted":"2026-08-12 14:32:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000720-fix-doc-publishing","meta_rev_id":"7c65cb4df1ab015bc23d0ff15e010e9f4340ef15","_number":1000720,"virtual_id_number":1000720,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-12 14:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-08-12 14:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-08-12 14:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-12 13:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d2038af4e7b1b656edb28269a0fb7119bca489d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-12 13:00:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00c83eb5630a63448e76dbdc8218cb44eb634cbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-12 13:01:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08eb1687c81a0329da4831b7be0c2ed081be8346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-12 13:35:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b223f81563fdb8b29af5cb2b8f565c7c22ce5dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 13:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c874bf845af4c0db1ab926405c5059e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6e20f8fee274cd88e5f315a66e0c960 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caeba4c29c7d4e1d9c62d3ba7beeb40d : SUCCESS in 4m 23s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6812db77fc944b8ca4920edc5b539787 : SUCCESS in 5m 46s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/0006172aa3394e7d82d3b786c96e0d40 : SUCCESS in 2m 54s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3986d1b50d0840a58a69ba27263dda72 : SUCCESS in 2m 52s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/fea53785316542d491bbf11c8b7d1f7f : SUCCESS in 3m 14s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2818e8433275446b8945a4e4d72c8ca7 : SUCCESS in 2m 53s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/5d29cc1f2af14ec69a09ebe5169662b2 : SUCCESS in 3m 25s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/9832d3040fc44b51ad54d638ed661b5a : SUCCESS in 3m 28s\n- rally-tox-zuul-ansible-lint https://zuul.opendev.org/t/openstack/build/f05234091ad24320986ba79f158b82fd : SUCCESS in 2m 13s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/17b579b4a2294a87b9c72ca32e4e2237 : SUCCESS in 3m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d7b6bc2488564703bdbbd8ca37c0c1af : SUCCESS in 6m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6208ae4db4cc4a29a0d4986d16fdc7c0 : SUCCESS in 5m 47s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e56c44af21d4e4d93c1ae6b81b07a50 : SUCCESS in 6m 30s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e51dc82dc2734a0582b516485886cdc1 : SUCCESS in 6m 12s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/ba6d326177954d5cafa8d9ec6a42079f : SUCCESS in 7m 37s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/0fc679b4a9f846869c8dc8c25acb55fd : SUCCESS in 16m 36s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/5179011d64e946ed9296a04236c86034 : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":3},{"id":"6e31fabaefa700c4acea5bb94ed65e0a52844ea4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-08-12 14:03:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d145e424b7540e6848823fef078cd4fbecc5e9c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:04:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9abf46e1d98c732eb46aaeb1fb62031f4e3f52d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:32:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a61a7d3688cc46ea9367e5cd24b2dc17\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fcf4631c27504d128b2aef5a09e39a04 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28813ee54df74c8eb224182754c4b381 : SUCCESS in 5m 58s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/557b85d6d1994289920bf63d17656929 : SUCCESS in 2m 53s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/cba13ee5978748ffad571de71110358b : SUCCESS in 3m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e863ce23a1f48f39f22ef016feda521 : SUCCESS in 2m 49s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/1160cc3effce4bca90aec6758b14bfac : SUCCESS in 2m 34s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/de3edfcce63548b3a84725b40bed15b7 : SUCCESS in 3m 07s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/a0bca3fe9b0240588a6226a037e2f3cc : SUCCESS in 4m 14s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/14e8727b24544164a66600441407d84c : SUCCESS in 3m 06s\n- rally-tox-zuul-ansible-lint https://zuul.opendev.org/t/openstack/build/394425c81faf4c14b3a34510e8c7f844 : SUCCESS in 2m 49s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/e27f7e4b804f451da9483351f86d58e5 : SUCCESS in 2m 22s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/633ed02ea7f14d59ab22468483952fee : SUCCESS in 6m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/78c1466722814d0e86be547558c3af5a : SUCCESS in 5m 19s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02f46517d08e4d7ba08cf187a4f9a598 : SUCCESS in 5m 34s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d59830c944432fb78dca83d641243e : SUCCESS in 5m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/4e6979bf7a7745dabf631fb1a67dadd2 : SUCCESS in 5m 56s\n- rally-task-simple-job https://zuul.opendev.org/t/openstack/build/276df7219b3c44ceb9f0abe1851c18ac : SUCCESS in 26m 59s\n- rally-verify-tempest https://zuul.opendev.org/t/openstack/build/de9cc6d4f44743febc7e0bf267525ff8 : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":3},{"id":"aa756f3369e5b70dd29b4b7a509c69e04d9ef7f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:32:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7c65cb4df1ab015bc23d0ff15e010e9f4340ef15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 14:34:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24feae6f12ae43e8a51a232ff4311cb4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77cd9ea831546c09ac50075e9f019e3 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0eb25961c9d301ab63eb6ef7a957cd5e9980f6b7","revisions":{"667b1d282c8cf5c6b3a3e00d4a61b4c209029200":{"kind":"REWORK","_number":1,"created":"2026-08-12 13:00:53.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/20/1000720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/1000720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/1000720/1"}}},"commit":{"parents":[{"commit":"45313c0c81b43d2a4ab1fc044b8569d0c3bedc03","subject":"Merge \"Fix outdated import paths in plugin documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45313c0c81b43d2a4ab1fc044b8569d0c3bedc03"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 12:49:36.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 13:00:45.000000000","tz":120},"subject":"[docs] Fix publishing triggers","message":"[docs] Fix publishing triggers\n\nReadthedocs trigger has been broken for ~7 months due to changes in\ncommon job (openstackci secret was rotated to dummy value and new value\nhas not been added)\n\ndocs.openstack.org has been broken for years - common template has been\nupdated resulting in not calling promote job at all\n\nChange-Id: I524e77bc6ce13825421b3bb7820a743b0dc8ecaf\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/667b1d282c8cf5c6b3a3e00d4a61b4c209029200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/667b1d282c8cf5c6b3a3e00d4a61b4c209029200"}]},"branch":"refs/heads/master"},"8ef67380c247b9cdae8ec182ccce13300d3ec71d":{"kind":"REWORK","_number":2,"created":"2026-08-12 13:01:33.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/20/1000720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/1000720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/1000720/2"}}},"commit":{"parents":[{"commit":"45313c0c81b43d2a4ab1fc044b8569d0c3bedc03","subject":"Merge \"Fix outdated import paths in plugin documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45313c0c81b43d2a4ab1fc044b8569d0c3bedc03"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 12:49:36.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 13:01:18.000000000","tz":120},"subject":"[docs] Fix publishing triggers","message":"[docs] Fix publishing triggers\n\nReadthedocs trigger has been broken for ~7 months due to changes in\ncommon job (openstackci secret was rotated to dummy value and new value\nhas not been added)\n\ndocs.openstack.org has been broken for years - common template has been\nupdated resulting in not calling promote job at all\n\nChange-Id: I524e77bc6ce13825421b3bb7820a743b0dc8ecaf\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8ef67380c247b9cdae8ec182ccce13300d3ec71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8ef67380c247b9cdae8ec182ccce13300d3ec71d"}]},"branch":"refs/heads/master"},"0eb25961c9d301ab63eb6ef7a957cd5e9980f6b7":{"kind":"REWORK","_number":3,"created":"2026-08-12 13:35:43.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/20/1000720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/20/1000720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/20/1000720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/20/1000720/3"}}},"commit":{"parents":[{"commit":"45313c0c81b43d2a4ab1fc044b8569d0c3bedc03","subject":"Merge \"Fix outdated import paths in plugin documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45313c0c81b43d2a4ab1fc044b8569d0c3bedc03"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 12:49:36.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-08-12 13:35:29.000000000","tz":120},"subject":"[docs] Fix publishing triggers","message":"[docs] Fix publishing triggers\n\nReadthedocs trigger has been broken for ~7 months due to changes in\ncommon job (openstackci secret was rotated to dummy value and new value\nhas not been added)\n\ndocs.openstack.org has been broken for years - common template has been\nupdated resulting in not calling promote job at all\n\nChange-Id: I524e77bc6ce13825421b3bb7820a743b0dc8ecaf\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0eb25961c9d301ab63eb6ef7a957cd5e9980f6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0eb25961c9d301ab63eb6ef7a957cd5e9980f6b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
