)]}'
{"id":"openstack%2Frally~1003312","triplet_id":"openstack%2Frally~master~If3505f703bb3dd991ce6396f29a4c33fb091e6b6","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-08 16:12:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3505f703bb3dd991ce6396f29a4c33fb091e6b6","subject":"Bump minimum PrettyTable","status":"MERGED","created":"2026-09-01 17:13:12.000000000","updated":"2026-09-08 16:13:53.000000000","submitted":"2026-09-08 16:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003312","meta_rev_id":"1f502df315398072d87fa03b3791b593963215b4","_number":1003312,"virtual_id_number":1003312,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 16:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-08 15:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-08 15:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 17:24:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 15:54:33.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"3508617b41dc29eb1c1108745a014982b6e333bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-01 17:13:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"986505636d915d274f93f536d156209cc6b60e94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f611735208994f8aa82e69d4bf57370c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c26d8ba4ca1476f8ccf30e4c18aa196 : SUCCESS in 4m 59s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6fc5e229f40c48c793b832c87d12e779 : SUCCESS in 4m 39s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f92a001acbb9478187669bb1f2eb1428 : SUCCESS in 3m 18s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d552d7c5eea14d05baf91b6bd9e3051a : SUCCESS in 2m 39s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/caf530bd150f4457aa99afff1ea9415d : SUCCESS in 3m 18s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/007e8de518b8464dbcd73390a068b95c : SUCCESS in 1m 59s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/f5d0e08704ee40fab14b53ac8e06c8a0 : SUCCESS in 1m 56s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/79bc863c88a44f849f7bc033806aef96 : SUCCESS in 3m 10s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/30d126e328464066b093f9f9433ac969 : SUCCESS in 2m 32s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/29acaf9ac6d6438194a5e0a3186df7fe : SUCCESS in 4m 43s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/1ce0cc76d35a45139899437ca622c9a4 : SUCCESS in 5m 47s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92a8121e0e8b497ba76b53d1a39df981 : SUCCESS in 5m 34s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdb0dfd8534f4027a8a52b9eb8b6ca69 : SUCCESS in 6m 05s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/654f3cc998464f2a822a3b9081cbb5ea : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b7109b93860844ce425733fb174a783ff2b43691","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-08 15:54:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c18cff954c81db54509efce898f5524b121a0df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:54:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4654585cf4f98b9ca206863de70e3aa08ff147d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:12:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea44118a64b64cd7b0d15252c31878cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c843cba1b514196b60b47a4e38174cc : SUCCESS in 4m 38s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/51e99a8d26264b03bb1dfe78d739cadd : SUCCESS in 4m 53s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/1c5acca32f0b4389bdd3807e84ef9d9c : SUCCESS in 2m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb429a840a84d168329209f101336dc : SUCCESS in 4m 03s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/1865e4dec0654f14b2ac4fcfd6218631 : SUCCESS in 4m 03s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2a74bc082a7a4ece9004727b3c697641 : SUCCESS in 2m 12s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/75194a66672d48578db0c363e8a6b666 : SUCCESS in 3m 11s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/a289fa24a54b403a9202de671b85c6f4 : SUCCESS in 2m 29s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b82f7bcaf87741b9bdaa5a0e61203f37 : SUCCESS in 3m 08s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3e7fe1ccb9c84495937fba01d2938663 : SUCCESS in 6m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4f245914b0514244a9593197d6f2f4e2 : SUCCESS in 6m 17s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f61f95e9b08e486b9fcf876894f75b2a : SUCCESS in 6m 02s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c073db9399754359a05779905040278d : SUCCESS in 6m 51s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/855bf0cc57ce4c30bde5e31650699c43 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"33e69a22b6c615e245d1b151e6f1e2aa917901ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:12:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f502df315398072d87fa03b3791b593963215b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:13:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51bbbc97fc2043dbad665ea3d003c8d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a872a69871884b068f668323bdbc4216 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b91f02493a35a9341f4ee55446eadfc252454568","revisions":{"b91f02493a35a9341f4ee55446eadfc252454568":{"kind":"REWORK","_number":1,"created":"2026-09-01 17:13:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/1003312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/12/1003312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/1003312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/1003312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/1003312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/12/1003312/1"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-01 17:12:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-01 17:12:19.000000000","tz":540},"subject":"Bump minimum PrettyTable","message":"Bump minimum PrettyTable\n\n... to remove the compabitility logic. PrettyTable 3.12.0 was released\non Oct 30, 2024.\n\nChange-Id: If3505f703bb3dd991ce6396f29a4c33fb091e6b6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b91f02493a35a9341f4ee55446eadfc252454568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b91f02493a35a9341f4ee55446eadfc252454568"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
