)]}'
{"id":"openstack%2Frally~1003846","triplet_id":"openstack%2Frally~master~I85cdf48c8c6e396203cf258d8e329ed0de9a2adf","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-09-08 15:55:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-08 16:12:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85cdf48c8c6e396203cf258d8e329ed0de9a2adf","subject":"Fix typo","status":"MERGED","created":"2026-09-03 15:50:52.000000000","updated":"2026-09-08 16:14:55.000000000","submitted":"2026-09-08 16:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003846","meta_rev_id":"3db952f4eface2143f0125461be6b65f10457ea0","_number":1003846,"virtual_id_number":1003846,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 16:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-08 15:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-08 15:55:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:53:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2026-09-03 16:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"628f947079000db74c91082cb8ebde6e52e55734","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-03 15:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"720ad50455f2b1f2bf570fde2e9a1c5af0597b12","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-03 15:53:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5bf62f18a62621d6ecc7fabde90d6977e129f92e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-03 15:53:16.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"ce1a9450ea863b08365ae03e975a85f5eb2c4dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:19:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b345dad2b59f400e8ca1e35dd715f85f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6b3c0e329d4f92a7b9740af69f24ca : SUCCESS in 5m 22s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0a431f32c3c8468ea06d7aceb79906b0 : SUCCESS in 5m 35s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/b5a7488342ef4193acfa069aef640f63 : SUCCESS in 4m 45s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/57bb6107d48140af9d7cc1c0cf27a2ae : SUCCESS in 2m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/aec88d7b6a3a447e8c4fa3f3bebecaad : SUCCESS in 4m 22s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/14a41ae527944696a5fab3de9176dd49 : SUCCESS in 3m 06s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/3459deef0df943abb5f1b558976695a8 : SUCCESS in 3m 41s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/b32308679d754a9594a8e71479973946 : SUCCESS in 2m 48s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/da89d15ff9ed40c496c08124d70940d2 : SUCCESS in 6m 52s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6967f9f3564542139d68a5ba61f67b12 : SUCCESS in 7m 08s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4878bf3fd7084b969a8b86ce0899856c : SUCCESS in 6m 21s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98d341d498844ee0abec0dcb88740718 : SUCCESS in 5m 50s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc196b1a56b044f0938e76656f6ee9d9 : SUCCESS in 6m 51s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/af16fa3911954233be0d70837919e4ff : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":2},{"id":"680fc19654b1c21974a23d3deb050927def4ac58","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-08 15:55:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"540c2e60cd340a7b2974fcdcdc63b98dd90aaaf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:55:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72fc4d674edf62f4b3643e526cd4fce94ed936b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:12:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/901a085ce7424e60ae663f6827367c44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2d13a43f7d74364a4ea1ef036ba9423 : SUCCESS in 4m 34s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/58213b2b167d4f2f9f09b8f954a1cf45 : SUCCESS in 4m 06s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/92461847ae3c4184bd429ff6d0d31c2f : SUCCESS in 2m 04s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4444e460db4a73a592c17244c6b391 : SUCCESS in 2m 54s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/6d9966d708ba471d88e2367eefb96fdc : SUCCESS in 3m 00s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/dedd4f8bb108492c9c4626acdb477c10 : SUCCESS in 2m 59s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/7702065673b3402a96c470ed29e65dd4 : SUCCESS in 2m 33s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/997c4290f1994cb5874f250a33dfa989 : SUCCESS in 2m 15s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/81379640ee69474fb4e4c76548acf275 : SUCCESS in 2m 10s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/e7fce0a523c24ac88db9b294db500af1 : SUCCESS in 7m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7bc511386a454103a4c2beca06ed26ff : SUCCESS in 8m 34s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/315c5280473e4db6ae1c21c332b93774 : SUCCESS in 6m 30s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f309493d9c21428e8388710d0aa1bc71 : SUCCESS in 6m 35s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/991fb31e19794ff1a98295d4fa78decf : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"fa48daecf4478f2d13078e1859f8170b4e0fdace","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:12:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3db952f4eface2143f0125461be6b65f10457ea0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:14:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1174b3d4f6f24ea580ffda36ff9a437c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46ff491be7d4693b163cc2c74798171 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b38702080b0ae72e68341d6d2266dddd9f3d05f0","revisions":{"ef54a23674b2288a00bd503e67b1ea7a3f5c7ae8":{"kind":"REWORK","_number":1,"created":"2026-09-03 15:50:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/46/1003846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/46/1003846/1"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-03 15:50:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-03 15:50:52.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I85cdf48c8c6e396203cf258d8e329ed0de9a2adf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ef54a23674b2288a00bd503e67b1ea7a3f5c7ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ef54a23674b2288a00bd503e67b1ea7a3f5c7ae8"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"b38702080b0ae72e68341d6d2266dddd9f3d05f0":{"kind":"REWORK","_number":2,"created":"2026-09-03 15:53:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/46/1003846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/46/1003846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/46/1003846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/46/1003846/2"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-03 15:50:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-03 15:53:04.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I85cdf48c8c6e396203cf258d8e329ed0de9a2adf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b38702080b0ae72e68341d6d2266dddd9f3d05f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b38702080b0ae72e68341d6d2266dddd9f3d05f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
