)]}'
{"id":"openstack%2Frally~1004227","triplet_id":"openstack%2Frally~master~Ib16abf92054ebfa4e2a96c1d6dc977076043cc3f","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-08 16:18:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib16abf92054ebfa4e2a96c1d6dc977076043cc3f","subject":"tox: Drop redundant passenv","status":"MERGED","created":"2026-09-05 18:18:03.000000000","updated":"2026-09-08 16:19:24.000000000","submitted":"2026-09-08 16:18:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004227","meta_rev_id":"4f463eb35bf4e94014de46d37158f8e00b32f0f4","_number":1004227,"virtual_id_number":1004227,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 16:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-08 15:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-08 15:55:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 18:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 15:55:38.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"5e5a7144b7fb9835af5e2525f0442db024374ee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d55bb7eb96ca32cd3f749a741280d2e3b9dff407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 18:52:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4bd8418f34f4c42b4844c4940eeffc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30101e4ee0a34d099ed49a37a573d7b0 : SUCCESS in 7m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/baf6201e539546568261632213b82ef2 : SUCCESS in 4m 48s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/6e404a270d564cc08232dc054499cc36 : SUCCESS in 3m 33s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fbe0041b7ae43ee99feb80e7ab00af1 : SUCCESS in 2m 34s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/45861331532f4b7d83d79a19c7537f26 : SUCCESS in 7m 16s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2f3076192b4b40eb9e734fae4e33859a : SUCCESS in 5m 06s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/67d5ff978c4e4ca8a10b5a12dc3aba53 : SUCCESS in 4m 42s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/08df7c4605474a87937985e1467c1293 : SUCCESS in 3m 17s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/68ef5bf30f53440c8907366028efdec8 : SUCCESS in 3m 18s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/475a0390ccf44e60953e10ed8caf0b48 : SUCCESS in 7m 24s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0a0574e8a3434d2f99c20be715cbb54b : SUCCESS in 8m 24s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7537286d871405b94a39816b4d3fedb : SUCCESS in 6m 43s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce748762bfda4c908e446ef3aa9563d6 : SUCCESS in 6m 31s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/246e965dd659431aa15193563aa222c1 : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d82872393abe4955b7d357c02452556e79362c53","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-08 15:55:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3b80fbcec0dcc95c0df2b1f42a7a8cef7d7a626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 15:56:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb84acb265e80091d2d825f35ae0110b1605d0b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:18:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eec5984274e475b9b62a3e0614ed99f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2033ca4404314da7910e7b5b8fcddff9 : SUCCESS in 6m 32s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/86ce4d2d834f454589abb364787ac372 : SUCCESS in 5m 36s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/7243c1592573490ca2f043945efb8581 : SUCCESS in 2m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e965ebef3be74b5880a83d3310c4cd9f : SUCCESS in 4m 03s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/2e23f1c8c9bd4adfbfbe0227fd8de20f : SUCCESS in 3m 00s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/37ce2c59810049ea8f06cc2fc1da1043 : SUCCESS in 5m 26s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/4697d7d994ae43cf94fa3ebe43741513 : SUCCESS in 3m 08s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/61794717c53f461cb60f13008f3587dd : SUCCESS in 3m 07s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/7933de16bf4d4a138c8c7ece2d835941 : SUCCESS in 2m 16s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/3a82d8ec9108474db22cd772c25e2c5b : SUCCESS in 7m 10s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6a16e392821e4081ad1185e1c7b72f92 : SUCCESS in 6m 18s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19c835a0d6ff41a2886c9ac4ba29cb38 : SUCCESS in 6m 11s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33865469406d496e95ed29be42032a1e : SUCCESS in 6m 30s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/fdae5b953a304fd78d804a463d1551bf : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"44c72d980f49f7ef9d9bfdfb3b54ea39c29ba39e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:18:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f463eb35bf4e94014de46d37158f8e00b32f0f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 16:19:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f109ef9b804eebba961bb7f9b70531\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4c592395f145a8af417ed454036068 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18f33e14167fc118f343f9be9afa1ec6b40aed4b","revisions":{"18f33e14167fc118f343f9be9afa1ec6b40aed4b":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:18:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/1004227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/27/1004227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/1004227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/1004227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/1004227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/27/1004227/1"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:17:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:17:54.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: Ib16abf92054ebfa4e2a96c1d6dc977076043cc3f\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/18f33e14167fc118f343f9be9afa1ec6b40aed4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/18f33e14167fc118f343f9be9afa1ec6b40aed4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
