)]}'
{"id":"openstack%2Frally~1004757","triplet_id":"openstack%2Frally~master~I2c86dbd0463b5f208eeba07604fef5406c0704d2","project":"openstack/rally","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 07:59:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c86dbd0463b5f208eeba07604fef5406c0704d2","subject":"Regenerate sample config file","status":"MERGED","created":"2026-09-09 02:58:42.000000000","updated":"2026-09-10 08:00:18.000000000","submitted":"2026-09-10 07:59:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004757","meta_rev_id":"93c642185dea1e2e7b20f87cb511b48570e64c2e","_number":1004757,"virtual_id_number":1004757,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 07:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-10 07:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-10 07:44:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 03:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:44:37.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"1b691c02e7a4e04cc2ca9170bfb8a4a76f873e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-09 02:58:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30fd71f360a87cb9bce2057142369b7c9ca4da46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 03:48:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0f8ac577f274b6dba572e4a18ab0eab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd01e6b246b44f69a5867110f7299e74 : SUCCESS in 5m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a2e70e0062924d7885361d0a4985a033 : SUCCESS in 3m 34s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/0ab6ec463a8d4a3d9dbcf22dcaf8b5e1 : SUCCESS in 4m 37s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fd0283f6b64f429d9f3c1f98047097 : SUCCESS in 2m 28s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/23f47bbfdea74e41a8281ab30012980b : SUCCESS in 4m 55s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/a4ba27f60b0945ce84fd681913f4db2e : SUCCESS in 3m 19s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ac15bf71e8774fcf9120663bd4db1df8 : SUCCESS in 3m 56s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/59686bf7634e4e73a1d213639af2ea80 : SUCCESS in 4m 25s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/0b0a4c1f10464fa698b6e3676b3e6155 : SUCCESS in 4m 27s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b44944b5c2ee43c6bf5295cb42cf4143 : SUCCESS in 8m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/fc31ec8800ee47c38a37ca63204f1cc5 : SUCCESS in 6m 37s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5724d9508759469c9b958ecbacdbceab : SUCCESS in 6m 43s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6fca0678aa24db4ae83d8d1bcc88118 : SUCCESS in 6m 08s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/9f94a59732cc489b84b312dcdee60324 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f72fb6bf5e42296a43f6b6695371f19591378a65","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-10 07:44:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f42921cf71d531d31eb9c0dd1d3be9c66443b8fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:44:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b4b129a3a651c763c7a135665ac2586c5230da4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:59:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e32e9a44d44442a2e78f5f9917ef36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badf150aad5d4f89893516878ef29fca : SUCCESS in 5m 39s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/9992b05e095a47f5b43277a9b38df08e : SUCCESS in 4m 19s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/a166c594edd04f14a89cc24a15d5c5ea : SUCCESS in 3m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/79bd5ceb12b04bb3b810ba8f0735ad63 : SUCCESS in 2m 56s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/a12948bfbada46a99c3107c9d42e059a : SUCCESS in 3m 50s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/3b661d0111d54678a34bc353c98f6952 : SUCCESS in 3m 16s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/7ccfff61ee004915b20eeddbffa53faa : SUCCESS in 3m 01s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/b8cc7045b72a4d50846084d37d77b786 : SUCCESS in 3m 11s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/a8d3cf2c3978437e9d3be37d435194e6 : SUCCESS in 2m 58s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/89e5cfc359a342cd8de947ea0902d10e : SUCCESS in 6m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/47aa4cbd60564c17b9ce0e25e740afc4 : SUCCESS in 6m 31s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3da4de62f495471e8fdc2787b752ebc2 : SUCCESS in 6m 08s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcde8706daed4612baddfa7410cb2c19 : SUCCESS in 6m 16s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/c997a5675398401fb9651d314bed6de1 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ab934adde34ade44a5991ad6c3642ddbe34aed32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 07:59:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"93c642185dea1e2e7b20f87cb511b48570e64c2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 08:00:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/386a3d73f4ce4673a0400784ed66405f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/138aa5fedd8e4d918389a2cec1a4c60d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b70ab4859ffbd2dec2d9336156e3a18f61569fa5","revisions":{"b70ab4859ffbd2dec2d9336156e3a18f61569fa5":{"kind":"REWORK","_number":1,"created":"2026-09-09 02:58:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/1004757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/57/1004757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/1004757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/1004757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/1004757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/57/1004757/1"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:58:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-09 02:58:33.000000000","tz":540},"subject":"Regenerate sample config file","message":"Regenerate sample config file\n\n... using the libraries in current upper constraints, mainly to clean\nup a few removed options.\n\nChange-Id: I2c86dbd0463b5f208eeba07604fef5406c0704d2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b70ab4859ffbd2dec2d9336156e3a18f61569fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b70ab4859ffbd2dec2d9336156e3a18f61569fa5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
