)]}'
{"id":"openstack%2Frally~1004952","triplet_id":"openstack%2Frally~master~Ie56c92d8acb60f474712ca5799aec79c4c2331f6","project":"openstack/rally","branch":"master","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-16 07:53:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"removed_from_attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-09-16 07:53:59.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"hashtags":[],"change_id":"Ie56c92d8acb60f474712ca5799aec79c4c2331f6","subject":"Remove Python 3.10 support","status":"NEW","created":"2026-09-10 02:26:15.000000000","updated":"2026-09-16 07:53:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fd0927c08b361d955225b457484ee92daa83b8e8","_number":1004952,"virtual_id_number":1004952,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 01:06:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2026-09-16 07:53:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2026-09-16 00:54:32.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 03:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-10 07:42:04.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"fded2237014a13f7bbc2595ec8926b4a81bfd253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:26:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1e4013013cc10e9b52fe7838e359ee96734a973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 03:11:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/232142afc94c4e22b9ff9e1875f6abf3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4593251a43ca4a0c809d7a1ad8e62280 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15d8b41e8d74536b60c5ee39204b7bd : SUCCESS in 6m 31s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/6564f1f5287046678a3fb7693e5ebb57 : SUCCESS in 6m 59s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/89596aba4dfb468ab02e68eeef65b5d2 : SUCCESS in 4m 21s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b0c0ed01a44913bc3384316b84238f : SUCCESS in 2m 36s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/e9076e7b5bbb4e2ba5a5e0ab109e88e0 : SUCCESS in 4m 21s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/cdcfd83ba3e44cd0857a137cecb2ba69 : SUCCESS in 3m 34s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/e69d84fd9c4143ae855004538fbe3852 : SUCCESS in 3m 48s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/64ca0546e46b4f2c9d4d13dac4ac209a : SUCCESS in 3m 47s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/ce8e8ad6b5eb474a953bc40a1cbf1e77 : SUCCESS in 7m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8df2978a3ae64b7bb7f1a09944ecb5d9 : SUCCESS in 7m 01s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf1937a8612f401abbce536675e798f3 : FAILURE in 2m 51s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8d203df2be444b1a19d25fe2e971d15 : SUCCESS in 7m 11s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/385182042f4e4d90ac854a2aba02163c : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c2668f809e90aa5e228e0793d974470ede701a4a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-10 07:42:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4cb4b1c5ba5af87efbaa41c452a2472c55febdc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 00:54:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d739892e4daafb2c0ec7c11e39ccd62c840420f7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 00:54:32.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2612ab6d1bcbbfe3f7164c6e4023d773c2992a80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 01:06:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85519e4b539a41a3b6289bf9beb745e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8faf3ffda6904d2694310f9ce9f908b7 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e3929dec6f04ae596995d0c5622cf52 : SUCCESS in 4m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e125392564214fe08aba833dbb51ec6b : SUCCESS in 4m 20s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/811496fc16ea4330bae005547098676a : SUCCESS in 3m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/daaf7063a31649c799d9161e6b79af38 : SUCCESS in 2m 43s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/74bb700d01bc494b8120656382b94c4b : SUCCESS in 3m 38s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/88802baf67e84c21aa4184a4682643c1 : SUCCESS in 3m 04s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/5939d25aef744ac1af4c44a33862d5aa : SUCCESS in 2m 04s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/544285a1dae6457f9b41c2e9ce6cf594 : SUCCESS in 3m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/127079b8fbfb4366b29e6473830a9e82 : SUCCESS in 6m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/28d76c2faf634ebfa2b1b93d76d7f9c7 : SUCCESS in 6m 27s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a87cc7aa0fa47af83981106c1356e1a : SUCCESS in 5m 20s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/2830db22980b4a1cb8aad50c80d3f537 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":2},{"id":"fd0927c08b361d955225b457484ee92daa83b8e8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-16 07:53:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3fec09d5875862c1aa7b08a19d5958e552bef3d6","revisions":{"0e63aecc267d5a03d4ae41f0d61fe9a660eb8390":{"kind":"REWORK","_number":1,"created":"2026-09-10 02:26:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1004952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/1004952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/1004952/1"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:25:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:26:07.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 was removed from tested runtimes during 2026.2 cycle.\n\nDepends-on: https://review.opendev.org/1004950\nChange-Id: Ie56c92d8acb60f474712ca5799aec79c4c2331f6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0e63aecc267d5a03d4ae41f0d61fe9a660eb8390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0e63aecc267d5a03d4ae41f0d61fe9a660eb8390"}]},"branch":"refs/heads/master"},"3fec09d5875862c1aa7b08a19d5958e552bef3d6":{"kind":"REWORK","_number":2,"created":"2026-09-16 00:54:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/1004952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/1004952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1004952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/1004952/2"}}},"commit":{"parents":[{"commit":"3790bd061e5e74bda4e1f7a5d635158eb3901331","subject":"Fix backward incompatible change of rally 5.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3790bd061e5e74bda4e1f7a5d635158eb3901331"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-10 02:25:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-16 00:54:15.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 was removed from tested runtimes during 2026.2 cycle.\n\nDepends-on: https://review.opendev.org/1004950\nChange-Id: Ie56c92d8acb60f474712ca5799aec79c4c2331f6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3fec09d5875862c1aa7b08a19d5958e552bef3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3fec09d5875862c1aa7b08a19d5958e552bef3d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
