)]}'
{"id":"openstack%2Frally~1005440","triplet_id":"openstack%2Frally~master~I6630a7bad1dde16a0788903637e6e7c9894e6daa","project":"openstack/rally","branch":"master","attention_set":{"9545":{"account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"last_update":"2026-09-15 02:16:36.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 02:16:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6630a7bad1dde16a0788903637e6e7c9894e6daa","subject":"Fix outdated project links","status":"NEW","created":"2026-09-13 17:45:57.000000000","updated":"2026-09-15 03:17:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0ed3d6618667acb55a4ed88cb00243bf49e10739","_number":1005440,"virtual_id_number":1005440,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 03:17:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-13 18:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 10:48:57.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"}],"messages":[{"id":"3689e1631cbd94576e9f076c424eacd5535db548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-13 17:45:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6add084b8201c0e293787249481da25b2725b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-13 18:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5757139fa8141cf942fb724248032a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeddceff190d4698bd2af44f4a775b89 : SUCCESS in 4m 29s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e78f8ef72bc247669881ab0445b5b5c9 : SUCCESS in 5m 22s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/7c769921ba96495ba222c7ed517a33d4 : SUCCESS in 3m 23s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa653120d3f742209a1a8ae9d2ea45d3 : SUCCESS in 2m 44s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/e2c0b76438cc45bdbea69b29801eab97 : SUCCESS in 4m 11s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/2335cad4e2154c60af06e9e894471865 : SUCCESS in 2m 13s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/0b0dd3c07778427a801d53752ada205f : SUCCESS in 3m 22s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/4fa92abaaba14b79825bcf824d1c0c21 : SUCCESS in 3m 10s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/51e8cdd9094b485fbb25d29d8e562a13 : SUCCESS in 3m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a6da4da7f8ed4019a0c55161c9652d54 : SUCCESS in 6m 13s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/d4d160c98717462495faa1fcd7ab81cd : SUCCESS in 5m 45s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fa8296858894af5ae93fd66a96d14b9 : SUCCESS in 9m 46s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73cff2dbe55a44b384fe845666ec57fe : SUCCESS in 6m 32s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/e453a2b6f2714e4a993eb8764d912c7c : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"617b5398121950b8c94a5290d9e5bfa0930482c1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-14 10:48:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2423e67a347bbd010c24f25dc3b43237cdeffb30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 02:16:36.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bef133f10405ee70225043cf6100551164e36af5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 02:16:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0ed3d6618667acb55a4ed88cb00243bf49e10739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 03:17:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8675d6dfb214c13889024a625a5d567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0722eda178014357ab318e030e75711b : SUCCESS in 7m 36s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/b323e9a9b63a4912bc3a884aa52ff036 : SUCCESS in 5m 38s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/5187bda52d9a41eda13e420015c0aa9a : SUCCESS in 4m 11s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec953c552e9f434b9341cc4b9069db64 : SUCCESS in 3m 14s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/586c9a61ba144c82a8bc0ee5cc90715e : SUCCESS in 4m 16s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/a7aa9f6d2bda46ceab555c0ff7500385 : SUCCESS in 3m 43s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/ab6c219f25324c029ef7a1b1b07dc8d5 : SUCCESS in 5m 08s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/b3cb7f0b4860462ca008234f19866a0d : SUCCESS in 3m 20s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/d605566a26134a42a9262cd4ad8cb5ce : SUCCESS in 2m 28s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/c3e413278c6544ad85e8129d11067962 : SUCCESS in 5m 15s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/75ded4f21f0f4aa8a8ba5c622f8e5099 : SUCCESS in 5m 57s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a98d63c6c5524eee9fd29ceeee1014cb : SUCCESS in 5m 38s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/465fad5662df4112aafec1139eb1d9c4 : SUCCESS in 5m 40s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/bffcc86dc3f34b28a0482dd990ccb618 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc5adfa4a06d9af1a2d2fd819d17986f388b975b","revisions":{"97a20cad05e3e1f99ab75e60a329ece61652df09":{"kind":"REWORK","_number":1,"created":"2026-09-13 17:45:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/1005440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/40/1005440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/40/1005440/1"}}},"commit":{"parents":[{"commit":"fbcb80e7ca0f4771066c6e9b226014942a08cdeb","subject":"Bring support for nested atomic actions to max_avg_duration_per_atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fbcb80e7ca0f4771066c6e9b226014942a08cdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:44:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:45:15.000000000","tz":540},"subject":"Fix outdated project links","message":"Fix outdated project links\n\nReplace or remove dead links.\n\nAlso use own repository instead of github repository, which is a mirror\ncurrently.\n\nChange-Id: I6630a7bad1dde16a0788903637e6e7c9894e6daa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/97a20cad05e3e1f99ab75e60a329ece61652df09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/97a20cad05e3e1f99ab75e60a329ece61652df09"}]},"branch":"refs/heads/master"},"fc5adfa4a06d9af1a2d2fd819d17986f388b975b":{"kind":"REWORK","_number":2,"created":"2026-09-15 02:16:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/1005440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/40/1005440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/40/1005440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/40/1005440/2"}}},"commit":{"parents":[{"commit":"fbcb80e7ca0f4771066c6e9b226014942a08cdeb","subject":"Bring support for nested atomic actions to max_avg_duration_per_atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fbcb80e7ca0f4771066c6e9b226014942a08cdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-13 17:44:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 02:16:35.000000000","tz":0},"subject":"Fix outdated project links","message":"Fix outdated project links\n\nReplace or remove dead links.\n\nAlso use own repository instead of github repository, which is a mirror\ncurrently.\n\nChange-Id: I6630a7bad1dde16a0788903637e6e7c9894e6daa\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fc5adfa4a06d9af1a2d2fd819d17986f388b975b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fc5adfa4a06d9af1a2d2fd819d17986f388b975b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
