)]}'
{"id":"openstack%2Frally~1005779","triplet_id":"openstack%2Frally~master~I74179c1617761e2c2e29ce78727da0e5db3f4e79","project":"openstack/rally","branch":"master","topic":"pep765","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-15 19:51:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9545\u003e replied on the change","reason_account":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I74179c1617761e2c2e29ce78727da0e5db3f4e79","subject":"Fix syntax error caused by return in a finally block","status":"NEW","created":"2026-09-15 18:27:41.000000000","updated":"2026-09-16 15:05:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c2a7c4658a5f9bbb94e1bd7488a47f14e71ee48d","_number":1005779,"virtual_id_number":1005779,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-15 18:43:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 18:43:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-15 19:51:22.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"CC"}],"messages":[{"id":"d88926bf86a5a7e0b4687fabd8abe7fa98735bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-15 18:27:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d33e7f88a762db1abd5c3701af3085be71178c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 18:43:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c4c8d43eb64a42bdbaa87921401825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92003eb953bb48cab80745c7340a9cdb : SUCCESS in 4m 32s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ae2ec6fffc304497807ae6f0bf3bcfa5 : SUCCESS in 6m 00s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/f1e1dedf65fb4050835126cfadb87008 : SUCCESS in 2m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f1b286a5ed4439953377e0d45afaf6 : SUCCESS in 2m 02s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/63ab168fd0074d4384db41183ca1340b : SUCCESS in 2m 21s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/7848d63ca51345acae796fb22cf9def3 : SUCCESS in 2m 39s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/d32c2395accb483da603e9069f5fba88 : SUCCESS in 2m 07s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/59b365411fc6471887a0438f3d5fe158 : SUCCESS in 4m 43s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/6c9e944a0498427580c8b94a0c79b820 : SUCCESS in 1m 45s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6c7bb790e6674767bdca71511cb0f806 : SUCCESS in 7m 44s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/8807f9c5ac2d46ada3c7b64e3074faa4 : SUCCESS in 7m 42s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53e0625e49154f6d8c9ec7c177ab05b2 : SUCCESS in 7m 20s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0c2dae9fc5c4c16a5870eba650273d5 : SUCCESS in 5m 47s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/8aba818c0a6d45079f1d438fb0792e1f : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"be0e4505c1ab8481f083612bfa303ad03a128878","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-15 19:51:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2a7c4658a5f9bbb94e1bd7488a47f14e71ee48d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-16 15:05:10.000000000","message":"Topic set to pep765","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7758253f0d0a81bc0c72d428d54166f27ae97c0","revisions":{"a7758253f0d0a81bc0c72d428d54166f27ae97c0":{"kind":"REWORK","_number":1,"created":"2026-09-15 18:27:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/1005779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/79/1005779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/1005779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/1005779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/79/1005779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/79/1005779/1"}}},"commit":{"parents":[{"commit":"fbcb80e7ca0f4771066c6e9b226014942a08cdeb","subject":"Bring support for nested atomic actions to max_avg_duration_per_atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/fbcb80e7ca0f4771066c6e9b226014942a08cdeb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 18:25:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-15 18:27:31.000000000","tz":540},"subject":"Fix syntax error caused by return in a finally block","message":"Fix syntax error caused by return in a finally block\n\nPEP765[1] is enforced in Python 3.14+ and this is strictly prohibited.\n\n[1]\nhttps://peps.python.org/pep-0765/\nPEP 765 – Disallow return/break/continue that exit a finally block\n\nChange-Id: I74179c1617761e2c2e29ce78727da0e5db3f4e79\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a7758253f0d0a81bc0c72d428d54166f27ae97c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a7758253f0d0a81bc0c72d428d54166f27ae97c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
