)]}'
{"id":"openstack%2Frally~1005952","triplet_id":"openstack%2Frally~master~I7f1a7f1869f29cec69f2a7fd8966b6313fd132ca","project":"openstack/rally","branch":"master","topic":"core-group","hashtags":[],"change_id":"I7f1a7f1869f29cec69f2a7fd8966b6313fd132ca","subject":"Update rally-core team","status":"MERGED","created":"2026-09-16 19:55:31.000000000","updated":"2026-09-16 20:42:50.000000000","submitted":"2026-09-16 20:41:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005952-core-group","meta_rev_id":"4aeb9099c9ac9000b5cca6a126e97abf70479711","_number":1005952,"virtual_id_number":1005952,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 20:41:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2026-09-16 19:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2026-09-16 19:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 20:21:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f83553f0ef9e129f94fd5403b21226c6b8f3574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-16 19:55:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a601107bbd9a6b30b60ca56a7fbac07a34081517","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2026-09-16 19:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88926cead94f286ed760a21115485fb9aad38404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:21:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15203463f8234e0698128f781fc0c8e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e62135375044396bfc6523eb1df162f : SUCCESS in 5m 06s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7dd13849d9b74686a7094270838e1330 : SUCCESS in 4m 56s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/b4432100949840f5868406c7c7aae95a : SUCCESS in 2m 38s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/991cb9feeae443a7b5b0ed8fbcea4330 : SUCCESS in 2m 40s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/afca6bbc74b24ee8a4c167678a79d1e2 : SUCCESS in 3m 08s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/4a278c754b534e929715e7902ef93476 : SUCCESS in 3m 05s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/6accb88fd41c40789ab5efb59d4eeff2 : SUCCESS in 2m 08s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/df2cdd14f0994864b335ebbf925626b4 : SUCCESS in 3m 39s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/35cb128371bd430e8e5355c2228292aa : SUCCESS in 2m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/20bbc9d3804e4f679b294dc74fc74134 : SUCCESS in 5m 34s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/6b58b6b3a0264eada6dc5206c5b23855 : SUCCESS in 6m 09s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25c940edbb2e4583808a1b3c2e776409 : SUCCESS in 5m 52s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b6b6d623af54a23bf49a4d490e6027b : SUCCESS in 7m 17s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/7302943cec3246a7a82d7c9cdb7e56d7 : SUCCESS in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d39e6cefdab79a67fb27e42c0850421f45435269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:21:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2cb299f35ca0baa5109bc1e08631c47357b923a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:41:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c3b06aa677c4f8bb120daf880fa657d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74ec011db9e4f9fbb23608c7fb87f33 : SUCCESS in 4m 38s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/d7e6f61781404aef87193a2f5e8133b5 : SUCCESS in 3m 26s\n- rally-tox-mypy https://zuul.opendev.org/t/openstack/build/1e13556194ef49c48c13f132b0fd17e4 : SUCCESS in 3m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f1786a3f93f4305847b9f3a3445e0de : SUCCESS in 2m 40s\n- rally-tox-py310 https://zuul.opendev.org/t/openstack/build/db98a75f7fe743be8449dc8ca463ea6c : SUCCESS in 3m 18s\n- rally-tox-py311 https://zuul.opendev.org/t/openstack/build/280f4691d3a84008909946c94ab6da92 : SUCCESS in 2m 43s\n- rally-tox-py312 https://zuul.opendev.org/t/openstack/build/89b7616929b84a7e84d99cb6d3006456 : SUCCESS in 2m 24s\n- rally-tox-py313 https://zuul.opendev.org/t/openstack/build/142d0d48ac7746b4a7471d178aa4df6a : SUCCESS in 4m 34s\n- rally-tox-samples https://zuul.opendev.org/t/openstack/build/b75f94f3a4f144c49b660aabe4553cc8 : SUCCESS in 2m 51s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5537530e5b1749799984c8c04681b2b0 : SUCCESS in 5m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9b8e01b7c265442ebcb7e782a02525d0 : SUCCESS in 6m 18s\n- rally-install-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d405a37de43452989b24e79764ba041 : SUCCESS in 7m 13s\n- rally-install-ubuntu-noble https://zuul.opendev.org/t/openstack/build/899051c3b88544c08ed1c5bbaaa2ec90 : SUCCESS in 7m 28s\n- rally-docker-build https://zuul.opendev.org/t/openstack/build/712fb42bcc734922b069bd0848b71a81 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ce0d923149469ca95a6aa1f85ecf3c773b056d3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:41:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4aeb9099c9ac9000b5cca6a126e97abf70479711","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 20:42:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ee12b5da9a3424a87c5c1fd18049746\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70883a69e7114d058ff1984b20bdeb8d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5776948ef932fcde62940ca7554b7e9016f5b1eb","revisions":{"5776948ef932fcde62940ca7554b7e9016f5b1eb":{"kind":"REWORK","_number":1,"created":"2026-09-16 19:55:31.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/52/1005952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/52/1005952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1005952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1005952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/52/1005952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/52/1005952/1"}}},"commit":{"parents":[{"commit":"06d66a9b0e44b5b3bbdca4685fd09d7a3387f00b","subject":"[docs] Rework plugin and CLI references, document envs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/06d66a9b0e44b5b3bbdca4685fd09d7a3387f00b"}]}],"author":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-16 19:52:39.000000000","tz":120},"committer":{"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","date":"2026-09-16 19:55:25.000000000","tz":120},"subject":"Update rally-core team","message":"Update rally-core team\n\nChange-Id: I7f1a7f1869f29cec69f2a7fd8966b6313fd132ca\nSigned-off-by: Andriy Kurilin \u003candr.kurilin@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5776948ef932fcde62940ca7554b7e9016f5b1eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5776948ef932fcde62940ca7554b7e9016f5b1eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
