)]}'
{"id":"openstack%2Frally~254797","triplet_id":"openstack%2Frally~master~Ib4bc836bbc47312a496df7f56b10dcc099fb6154","project":"openstack/rally","branch":"master","topic":"verify_export","hashtags":[],"change_id":"Ib4bc836bbc47312a496df7f56b10dcc099fb6154","subject":"[WIP]Add verify export command","status":"NEW","created":"2015-12-08 14:48:29.000000000","updated":"2017-12-18 05:29:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6379d6eb72119cda4ce79e1eb757916fe660809a","_number":254797,"virtual_id_number":254797,"owner":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"all":[{"value":-1,"date":"2015-12-10 15:34:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-12-10 15:34:01.000000000","updated_by":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"reviewer":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"state":"REVIEWER"}],"messages":[{"id":"83ed4aa11f0c628761b51370839fdf7404c2ddb4","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-08 14:48:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab80eda4e4dd021c9143f3eb17f09ee953f2527d","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-08 14:55:47.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"35370176e4b4f39a64bb93b239e3d9cf961feeec","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-08 15:20:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/docker-check/ : SUCCESS in 9m 58s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/rally-pg-py27-unit/ : FAILURE in 4m 23s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/fuel-community-6.0/ : FAILURE in 6m 16s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/rally-dsvm-pg-py27-cli/ : FAILURE in 26m 25s\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/rally-dsvm-pg-py34-cli/ : FAILURE in 31m 55s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/mos-7.0/ : FAILURE in 8m 7s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/mos-6.1/ : FAILURE in 8m 39s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTrjEhAIRFKM/rally-dsvm-sahara/ : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fc71d7f84fcf2a936b8281868104c46ee82394fa","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-08 15:32:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/docker-check/ : SUCCESS in 16m 3s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/rally-pg-py27-unit/ : FAILURE in 11m 22s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/fuel-community-6.0/ : FAILURE in 13m 4s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/rally-dsvm-pg-py27-cli/ : FAILURE in 30m 31s\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/rally-dsvm-pg-py34-cli/ : FAILURE in 33m 52s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/mos-7.0/ : FAILURE in 14m 57s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/mos-6.1/ : FAILURE in 15m 26s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTPbJdvpSjfm/rally-dsvm-sahara/ : FAILURE in 36m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ea8f0b67e08cb6cb162a62569ceb271f479bae16","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-08 15:59:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"55fe129e307800c9764b8c7608c8d599a178f924","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-08 17:17:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/docker-check/ : SUCCESS in 17m 45s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/rally-pg-py27-unit/ : FAILURE in 10m 29s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/fuel-community-6.0/ : SUCCESS in 17m 10s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/rally-dsvm-pg-py27-cli/ : SUCCESS in 1h 17m 56s\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/rally-dsvm-pg-py34-cli/ : FAILURE in 42m 13s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/mos-7.0/ : SUCCESS in 51m 46s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/mos-6.1/ : SUCCESS in 53m 14s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTawFzepXDRl/rally-dsvm-sahara/ : SUCCESS in 1h 7m 19s","accounts_in_message":[],"_revision_number":3},{"id":"c8f97ad9c1230ed340e1e04e5b3f8502480e655e","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-08 17:26:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1e4ec5ff4f29fea20a23c69c3b287f0de6230bc3","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-08 18:06:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/docker-check/ : SUCCESS in 5m 17s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/rally-pg-py27-unit/ : FAILURE in 2m 23s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/fuel-community-6.0/ : SUCCESS in 6m 35s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/rally-dsvm-pg-py27-cli/ : SUCCESS in 39m 59s\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/rally-dsvm-pg-py34-cli/ : FAILURE in 20m 10s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/mos-7.0/ : SUCCESS in 22m 44s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/mos-6.1/ : SUCCESS in 22m 32s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTAfQBXUbOdv/rally-dsvm-sahara/ : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":4},{"id":"5ada72d177a0dc01fa165e0458060f7b5515634b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-08 20:38:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-rally-pep8 http://logs.openstack.org/97/254797/4/check/gate-rally-pep8/7baf691/ : FAILURE in 2m 06s\n- gate-rally-docs http://docs-draft.openstack.org/97/254797/4/check/gate-rally-docs/b01613d//doc/build/html/ : SUCCESS in 2m 00s\n- gate-rally-python27 http://logs.openstack.org/97/254797/4/check/gate-rally-python27/348dad0/ : FAILURE in 1m 58s\n- gate-rally-python34 http://logs.openstack.org/97/254797/4/check/gate-rally-python34/67749da/ : FAILURE in 2m 55s\n- rally-coverage http://logs.openstack.org/97/254797/4/check/rally-coverage/d8f372e/ : FAILURE in 3m 50s\n- gate-rally-install-bare-precise http://logs.openstack.org/97/254797/4/check/gate-rally-install-bare-precise/44768de/ : SUCCESS in 3m 28s\n- gate-rally-dsvm-rally http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-rally/877dbdf/ : SUCCESS in 52m 15s\n- gate-rally-dsvm-rally-cinder http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-rally-cinder/1714508/ : SUCCESS in 46m 04s (non-voting)\n- gate-rally-dsvm-rally-heat http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-rally-heat/fc8e5a0/ : SUCCESS in 40m 59s (non-voting)\n- gate-rally-dsvm-rally-nova http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-rally-nova/075f3dd/ : SUCCESS in 53m 37s (non-voting)\n- gate-rally-dsvm-manila-multibackend http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-manila-multibackend/4719fd8/ : SUCCESS in 40m 43s (non-voting)\n- gate-rally-dsvm-manila-multibackend-no-ss http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-manila-multibackend-no-ss/3ea054b/ : SUCCESS in 34m 48s (non-voting)\n- gate-rally-dsvm-neutron-rally http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-neutron-rally/ca3c593/ : SUCCESS in 52m 31s\n- gate-rally-dsvm-neutron-unstable http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-neutron-unstable/9335232/ : FAILURE in 47m 30s (non-voting)\n- gate-rally-dsvm-cli http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-cli/7c139ac/ : SUCCESS in 45m 55s\n- gate-rally-dsvm-verify-light http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-verify-light/2834d9d/ : SUCCESS in 1h 04m 27s\n- gate-rally-dsvm-verify-full http://logs.openstack.org/97/254797/4/check/gate-rally-dsvm-verify-full/bd57465/ : SUCCESS in 1h 15m 31s\n- gate-rally-tox-self http://logs.openstack.org/97/254797/4/check/gate-rally-tox-self/40dbf08/ : FAILURE in 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d85cb650cc3ab3d8c02876314a84417aaab9c8d","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-09 14:26:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7879712d95613bc3a6c7ee153ece2339dd49184d","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-09 15:05:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/docker-check/ : SUCCESS in 5m 25s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/rally-pg-py27-unit/ : FAILURE in 2m 55s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/fuel-community-6.0/ : FAILURE in 14m 1s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/rally-dsvm-pg-py27-cli/ : SUCCESS in 39m\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/rally-dsvm-pg-py34-cli/ : FAILURE in 20m 3s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/mos-7.0/ : SUCCESS in 22m 38s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/mos-6.1/ : SUCCESS in 21m 33s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTXqBNwHWfVr/rally-dsvm-sahara/ : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":5},{"id":"2e16935107902e970bab83a839feb7b3465bf415","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-09 15:45:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-rally-pep8 http://logs.openstack.org/97/254797/5/check/gate-rally-pep8/e976791/ : FAILURE in 1m 43s\n- gate-rally-docs http://docs-draft.openstack.org/97/254797/5/check/gate-rally-docs/5438233//doc/build/html/ : SUCCESS in 1m 37s\n- gate-rally-python27 http://logs.openstack.org/97/254797/5/check/gate-rally-python27/8eb9205/ : FAILURE in 1m 32s\n- gate-rally-python34 http://logs.openstack.org/97/254797/5/check/gate-rally-python34/1f90254/ : FAILURE in 3m 17s\n- rally-coverage http://logs.openstack.org/97/254797/5/check/rally-coverage/1db59e0/ : FAILURE in 3m 20s\n- gate-rally-install-bare-precise http://logs.openstack.org/97/254797/5/check/gate-rally-install-bare-precise/e92b39a/ : SUCCESS in 3m 33s\n- gate-rally-dsvm-rally http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-rally/d1867e3/ : SUCCESS in 42m 29s\n- gate-rally-dsvm-rally-cinder http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-rally-cinder/6352102/ : SUCCESS in 48m 01s (non-voting)\n- gate-rally-dsvm-rally-heat http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-rally-heat/e5113cf/ : SUCCESS in 39m 25s (non-voting)\n- gate-rally-dsvm-rally-nova http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-rally-nova/71ce913/ : SUCCESS in 49m 11s (non-voting)\n- gate-rally-dsvm-manila-multibackend http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-manila-multibackend/a94f732/ : SUCCESS in 26m 02s (non-voting)\n- gate-rally-dsvm-manila-multibackend-no-ss http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-manila-multibackend-no-ss/396848c/ : SUCCESS in 41m 01s (non-voting)\n- gate-rally-dsvm-neutron-rally http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-neutron-rally/9479f36/ : SUCCESS in 49m 40s\n- gate-rally-dsvm-neutron-unstable http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-neutron-unstable/b2da9b9/ : FAILURE in 31m 08s (non-voting)\n- gate-rally-dsvm-cli http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-cli/214e563/ : SUCCESS in 42m 45s\n- gate-rally-dsvm-verify-light http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-verify-light/ebf552d/ : SUCCESS in 52m 05s\n- gate-rally-dsvm-verify-full http://logs.openstack.org/97/254797/5/check/gate-rally-dsvm-verify-full/2f360f8/ : SUCCESS in 1h 03m 00s\n- gate-rally-tox-self http://logs.openstack.org/97/254797/5/check/gate-rally-tox-self/1840d33/ : FAILURE in 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b7ada185bd7fe4ed05906aab69b405522a53d47","author":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"date":"2015-12-10 14:51:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6deb194bfc943578e813952853dd89e453f75a63","author":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"},"date":"2015-12-10 15:34:01.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed by RallyCI. Use `rally-ci recheck` to recheck\n\nSee status at: https://rally-ci.tk/\n- docker-check https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/docker-check/ : SUCCESS in 5m 18s\n- rally-pg-py27-unit https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/rally-pg-py27-unit/ : FAILURE in 2m 58s\n- fuel-community-6.0 https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/fuel-community-6.0/ : SUCCESS in 6m 29s\n- rally-dsvm-pg-py27-cli https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/rally-dsvm-pg-py27-cli/ : SUCCESS in 42m 17s\n- rally-dsvm-pg-py34-cli https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/rally-dsvm-pg-py34-cli/ : FAILURE in 22m 1s\n- mos-7.0 https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/mos-7.0/ : SUCCESS in 23m 16s\n- mos-6.1 https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/mos-6.1/ : SUCCESS in 22m 13s\n- rally-dsvm-sahara https://rally-ci.tk/logs/_rci_EVNTvAFKjtbsqe/rally-dsvm-sahara/ : SUCCESS in 38m 22s","accounts_in_message":[],"_revision_number":6},{"id":"e5c18f6512ea6d71fd02bc6792ce4c94d2e4ba66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-12-10 16:38:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-rally-pep8 http://logs.openstack.org/97/254797/6/check/gate-rally-pep8/f1836d9/ : FAILURE in 1m 30s\n- gate-rally-docs http://docs-draft.openstack.org/97/254797/6/check/gate-rally-docs/872c3db//doc/build/html/ : SUCCESS in 1m 28s\n- gate-rally-python27 http://logs.openstack.org/97/254797/6/check/gate-rally-python27/c59e32a/ : FAILURE in 1m 32s\n- gate-rally-python34 http://logs.openstack.org/97/254797/6/check/gate-rally-python34/2af7cd9/ : FAILURE in 3m 15s\n- rally-coverage http://logs.openstack.org/97/254797/6/check/rally-coverage/dfccea7/ : FAILURE in 2m 52s\n- gate-rally-install-bare-precise http://logs.openstack.org/97/254797/6/check/gate-rally-install-bare-precise/95f038b/ : SUCCESS in 4m 44s\n- gate-rally-dsvm-rally http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-rally/26b6eb5/ : SUCCESS in 36m 22s\n- gate-rally-dsvm-rally-cinder http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-rally-cinder/a9a31a0/ : SUCCESS in 48m 37s (non-voting)\n- gate-rally-dsvm-rally-heat http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-rally-heat/0104f01/ : SUCCESS in 38m 23s (non-voting)\n- gate-rally-dsvm-rally-nova http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-rally-nova/3331506/ : FAILURE in 1h 04m 36s (non-voting)\n- gate-rally-dsvm-manila-multibackend http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-manila-multibackend/759c2a4/ : SUCCESS in 43m 47s (non-voting)\n- gate-rally-dsvm-manila-multibackend-no-ss http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-manila-multibackend-no-ss/793e4fa/ : FAILURE in 38m 43s (non-voting)\n- gate-rally-dsvm-neutron-rally http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-neutron-rally/730e242/ : SUCCESS in 51m 05s\n- gate-rally-dsvm-neutron-unstable http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-neutron-unstable/7a5f7b1/ : FAILURE in 33m 57s (non-voting)\n- gate-rally-dsvm-cli http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-cli/55634ff/ : SUCCESS in 49m 42s\n- gate-rally-dsvm-verify-light http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-verify-light/9ba3fef/ : SUCCESS in 55m 04s\n- gate-rally-dsvm-verify-full http://logs.openstack.org/97/254797/6/check/gate-rally-dsvm-verify-full/a59d0cd/ : SUCCESS in 1h 08m 36s\n- gate-rally-tox-self http://logs.openstack.org/97/254797/6/check/gate-rally-tox-self/1ad2f8d/ : FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c0d5c0e43cfa753f09045ccb6b017444d9e98a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 05:29:29.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c86b79da681f15c170b4c2958a6fa997587b926e","revisions":{"bbe6bd4194b106bce97b81bf4264f4db2a682f73":{"kind":"REWORK","_number":1,"created":"2015-12-08 14:48:29.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/1"}}},"commit":{"parents":[{"commit":"8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7","subject":"[gates] move ceilometer alarts scenarios to non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:47:20.000000000","tz":120},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t revew it. Don\u0027t waste you time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bbe6bd4194b106bce97b81bf4264f4db2a682f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bbe6bd4194b106bce97b81bf4264f4db2a682f73"}]},"branch":"refs/heads/master"},"e3c1265b52e4ff5a19fb740fb39d0495d7279054":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-12-08 14:55:47.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/2"}}},"commit":{"parents":[{"commit":"8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7","subject":"[gates] move ceilometer alarts scenarios to non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:55:47.000000000","tz":0},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t review it. Don\u0027t waste your time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e3c1265b52e4ff5a19fb740fb39d0495d7279054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/e3c1265b52e4ff5a19fb740fb39d0495d7279054"}]},"branch":"refs/heads/master"},"b9e1e563bdd0e2b3386f166f74d564d60eddc224":{"kind":"REWORK","_number":3,"created":"2015-12-08 15:59:08.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/3"}}},"commit":{"parents":[{"commit":"8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7","subject":"[gates] move ceilometer alarts scenarios to non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 15:58:42.000000000","tz":120},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t review it. Don\u0027t waste your time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b9e1e563bdd0e2b3386f166f74d564d60eddc224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b9e1e563bdd0e2b3386f166f74d564d60eddc224"}]},"branch":"refs/heads/master"},"cbd58b57a8086f986345732ccfd33b0bd580f184":{"kind":"REWORK","_number":4,"created":"2015-12-08 17:26:36.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/4"}}},"commit":{"parents":[{"commit":"8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7","subject":"[gates] move ceilometer alarts scenarios to non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 17:26:24.000000000","tz":120},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t review it. Don\u0027t waste your time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cbd58b57a8086f986345732ccfd33b0bd580f184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cbd58b57a8086f986345732ccfd33b0bd580f184"}]},"branch":"refs/heads/master"},"13083d17ebb385a896a551c47ebb8bf5f8f2368b":{"kind":"REWORK","_number":5,"created":"2015-12-09 14:26:50.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/5"}}},"commit":{"parents":[{"commit":"8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7","subject":"[gates] move ceilometer alarts scenarios to non-voting job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d50b3e04636430bdcb2f28e4ac7a7e5b352fac7"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-09 14:26:40.000000000","tz":120},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t review it. Don\u0027t waste your time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/13083d17ebb385a896a551c47ebb8bf5f8f2368b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/13083d17ebb385a896a551c47ebb8bf5f8f2368b"}]},"branch":"refs/heads/master"},"c86b79da681f15c170b4c2958a6fa997587b926e":{"kind":"REWORK","_number":6,"created":"2015-12-10 14:51:41.000000000","uploader":{"_account_id":12395,"name":"Roman Vasylets","email":"pomeo92@gmail.com","username":"rvasilets"},"ref":"refs/changes/97/254797/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/254797/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/254797/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/254797/6"}}},"commit":{"parents":[{"commit":"3e1520f834a99b8ea8e0986ed661f4bd3cd83150","subject":"Merge \"[StreamingAlgorithms] Unify exception type raised by result()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3e1520f834a99b8ea8e0986ed661f4bd3cd83150"}]}],"author":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-08 14:46:48.000000000","tz":120},"committer":{"name":"Roman Vasilets","email":"rvasilets@mirantis.com","date":"2015-12-10 14:51:26.000000000","tz":120},"subject":"[WIP]Add verify export command","message":"[WIP]Add verify export command\n\nSuper WIP! Don\u0027t review it. Don\u0027t waste your time.\n\nChange-Id: Ib4bc836bbc47312a496df7f56b10dcc099fb6154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c86b79da681f15c170b4c2958a6fa997587b926e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c86b79da681f15c170b4c2958a6fa997587b926e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":14817,"name":"Mirantis Rally CI","email":"rally-ci@mirantis.com","username":"ci50"}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
