)]}'
{"id":"openstack%2Frally~566949","triplet_id":"openstack%2Frally~master~Ife8cfb886c97fc8712bec8445c4c3241a98ccbd5","project":"openstack/rally","branch":"master","topic":"new_hacking","hashtags":[],"change_id":"Ife8cfb886c97fc8712bec8445c4c3241a98ccbd5","subject":"Ignore several hacking rule","status":"MERGED","created":"2018-05-08 17:10:49.000000000","updated":"2018-05-08 19:03:15.000000000","submitted":"2018-05-08 19:03:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"566949-1525806195707-0f824b7e","meta_rev_id":"0005bdf35f1084123e597d0142290c1a299d8914","_number":566949,"virtual_id_number":566949,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":2,"date":"2018-05-08 19:03:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"all":[{"value":2,"date":"2018-05-08 18:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"all":[{"value":1,"date":"2018-05-08 18:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-08 18:44:45.000000000","updated_by":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"reviewer":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"state":"REVIEWER"},{"updated":"2018-05-08 19:03:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91ecfe727c8ab178a97f4df67d8447020b2f33f5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2018-05-08 17:10:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ed6f577f84bd0c66dfe50f45d399c1864375fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 18:03:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/49/566949/1/check/openstack-tox-pep8/1daf74d/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/49/566949/1/check/openstack-tox-py27/ac54351/ : SUCCESS in 4m 39s\n- openstack-tox-py35 http://logs.openstack.org/49/566949/1/check/openstack-tox-py35/054095a/ : SUCCESS in 6m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/49/566949/1/check/build-openstack-sphinx-docs/0db58e6/html/ : SUCCESS in 5m 57s\n- openstack-tox-cover http://logs.openstack.org/49/566949/1/check/openstack-tox-cover/7bd72b3/cover/ : SUCCESS in 5m 55s\n- rally-database-migration http://logs.openstack.org/49/566949/1/check/rally-database-migration/0ad8829/ : SUCCESS in 8m 33s\n- rally-install-ubuntu-xenial http://logs.openstack.org/49/566949/1/check/rally-install-ubuntu-xenial/847269b/ : SUCCESS in 6m 50s\n- rally-install-centos-7 http://logs.openstack.org/49/566949/1/check/rally-install-centos-7/e5d08d5/ : SUCCESS in 8m 13s\n- openstack-tox-functional http://logs.openstack.org/49/566949/1/check/openstack-tox-functional/2ce0abd/ : SUCCESS in 6m 13s\n- rally-tox-samples http://logs.openstack.org/49/566949/1/check/rally-tox-samples/053e4ca/ : SUCCESS in 3m 54s\n- rally-tox-self http://logs.openstack.org/49/566949/1/check/rally-tox-self/184d8af/ : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9285226d7d46cecb519bd1b0768fad2603ce3c54","author":{"_account_id":6172,"name":"Boris Pavlovic","email":"boris@pavlovic.me","username":"boris-42"},"date":"2018-05-08 18:44:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2e81f4bfd470281fc163f83a9e6c0804104fd26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 18:44:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c8afc082e75f0fd9b1b9619c7e6faa94f78eaac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 19:03:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21c9909e89faba010a2e04dbccf0d12daec3586e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-08 19:03:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/49/566949/1/gate/openstack-tox-pep8/f2748d7/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/49/566949/1/gate/openstack-tox-py27/a97e037/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/49/566949/1/gate/openstack-tox-py35/480c564/ : SUCCESS in 4m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/49/566949/1/gate/build-openstack-sphinx-docs/4e27664/html/ : SUCCESS in 5m 13s\n- rally-database-migration http://logs.openstack.org/49/566949/1/gate/rally-database-migration/6470902/ : SUCCESS in 9m 06s\n- rally-install-ubuntu-xenial http://logs.openstack.org/49/566949/1/gate/rally-install-ubuntu-xenial/ea93dc5/ : SUCCESS in 6m 51s\n- rally-install-centos-7 http://logs.openstack.org/49/566949/1/gate/rally-install-centos-7/2c2618e/ : SUCCESS in 9m 36s\n- openstack-tox-functional http://logs.openstack.org/49/566949/1/gate/openstack-tox-functional/e6601bd/ : SUCCESS in 6m 41s\n- rally-tox-samples http://logs.openstack.org/49/566949/1/gate/rally-tox-samples/775b082/ : SUCCESS in 4m 05s\n- rally-tox-self http://logs.openstack.org/49/566949/1/gate/rally-tox-self/deae9fe/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41e633bac26040d0bdd65a670692964af9699fc4","revisions":{"41e633bac26040d0bdd65a670692964af9699fc4":{"kind":"REWORK","_number":1,"created":"2018-05-08 17:10:49.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/49/566949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/49/566949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/566949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/566949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/49/566949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/49/566949/1"}}},"commit":{"parents":[{"commit":"6237d9ae0e6f42c04630934760b5f581925473f7","subject":"Merge \"[version] Import rally.common.db.schema when necessary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6237d9ae0e6f42c04630934760b5f581925473f7"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2018-05-08 17:09:26.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2018-05-08 17:09:26.000000000","tz":180},"subject":"Ignore several hacking rule","message":"Ignore several hacking rule\n\nThe new hacking is released which made over \"pep8\" check fail.\nI do not find this checks useful and even more, it blocks us to make\nrelease. So let\u0027s just ignore them.\n\n  E731 do not assign a lambda expression, use a def\n  W503 line break before binary operator\n\nChange-Id: Ife8cfb886c97fc8712bec8445c4c3241a98ccbd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41e633bac26040d0bdd65a670692964af9699fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/41e633bac26040d0bdd65a670692964af9699fc4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6237d9ae0e6f42c04630934760b5f581925473f7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
