)]}'
{"id":"openstack%2Frally~653927","triplet_id":"openstack%2Frally~master~Iaa8182c010923eaf052221dcb08b1bd95f7d4ea8","project":"openstack/rally","branch":"master","topic":"requirements","hashtags":[],"change_id":"Iaa8182c010923eaf052221dcb08b1bd95f7d4ea8","subject":"update upper-constraints.txt for Sphinx","status":"MERGED","created":"2019-04-19 08:43:57.000000000","updated":"2019-04-20 18:39:15.000000000","submitted":"2019-04-20 18:39:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653927-1555785555436-b16e9fe7","meta_rev_id":"1b683021f9c1a7be88296c060d2853c0e8e3d6a9","_number":653927,"virtual_id_number":653927,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-04-20 18:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-04-20 18:22:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-04-20 18:22:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 18:22:01.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-04-20 18:39:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24e8443def3fc930d6838dc8add4aa9a37d67435","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-04-19 08:43:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37f3f92de496176aa8a4ea8bf2b2239a892a8dfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 09:20:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/653927/1/check/openstack-tox-docs/f13b21b/html/ : SUCCESS in 3m 01s\n- rally-tox-docs http://logs.openstack.org/27/653927/1/check/rally-tox-docs/cb9ceba/ : SUCCESS in 3m 06s\n- rally-tox-pep8 http://logs.openstack.org/27/653927/1/check/rally-tox-pep8/79e504f/ : SUCCESS in 3m 25s\n- rally-tox-py27 http://logs.openstack.org/27/653927/1/check/rally-tox-py27/e7831af/ : SUCCESS in 3m 24s\n- rally-tox-py34 http://logs.openstack.org/27/653927/1/check/rally-tox-py34/f59d19a/ : FAILURE in 4m 54s\n- rally-tox-py35 http://logs.openstack.org/27/653927/1/check/rally-tox-py35/b62a2ea/ : SUCCESS in 5m 23s\n- rally-tox-py36 http://logs.openstack.org/27/653927/1/check/rally-tox-py36/0bd1eb8/ : SUCCESS in 3m 19s\n- rally-tox-py37 http://logs.openstack.org/27/653927/1/check/rally-tox-py37/5111df1/ : SUCCESS in 4m 31s\n- rally-tox-cover http://logs.openstack.org/27/653927/1/check/rally-tox-cover/1aa1e7d/ : SUCCESS in 3m 49s\n- rally-tox-functional http://logs.openstack.org/27/653927/1/check/rally-tox-functional/f8fe1a8/ : SUCCESS in 7m 12s\n- rally-tox-functional-py3 http://logs.openstack.org/27/653927/1/check/rally-tox-functional-py3/1d4709e/ : SUCCESS in 5m 35s\n- rally-database-migration http://logs.openstack.org/27/653927/1/check/rally-database-migration/3512a04/ : SUCCESS in 6m 54s\n- rally-install-ubuntu-xenial http://logs.openstack.org/27/653927/1/check/rally-install-ubuntu-xenial/322d826/ : SUCCESS in 5m 09s\n- rally-install-ubuntu-bionic http://logs.openstack.org/27/653927/1/check/rally-install-ubuntu-bionic/1536da8/ : SUCCESS in 3m 35s\n- rally-install-centos-7 http://logs.openstack.org/27/653927/1/check/rally-install-centos-7/debd2f4/ : SUCCESS in 5m 58s\n- rally-tox-self http://logs.openstack.org/27/653927/1/check/rally-tox-self/4e8dd4d/ : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d3d3888cb99602ddbae099f36ef5b01d7db3b9aa","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-04-20 03:53:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4411a570bd04b166665398b7646fc55b289dae75","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-04-20 03:53:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ff44e1cc50822fbb58e96e26f2547a161a4eb243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 04:05:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/653927/3/check/openstack-tox-docs/f58adc1/html/ : SUCCESS in 3m 24s\n- rally-tox-docs http://logs.openstack.org/27/653927/3/check/rally-tox-docs/417ba41/ : SUCCESS in 4m 23s\n- rally-tox-pep8 http://logs.openstack.org/27/653927/3/check/rally-tox-pep8/57ebded/ : SUCCESS in 3m 37s\n- rally-tox-py27 http://logs.openstack.org/27/653927/3/check/rally-tox-py27/1b0d337/ : SUCCESS in 4m 30s\n- rally-tox-py34 http://logs.openstack.org/27/653927/3/check/rally-tox-py34/1c72b9f/ : SUCCESS in 5m 20s\n- rally-tox-py35 http://logs.openstack.org/27/653927/3/check/rally-tox-py35/1db6f86/ : SUCCESS in 4m 32s\n- rally-tox-py36 http://logs.openstack.org/27/653927/3/check/rally-tox-py36/38e7c8e/ : SUCCESS in 3m 12s\n- rally-tox-py37 http://logs.openstack.org/27/653927/3/check/rally-tox-py37/c8e0358/ : SUCCESS in 4m 00s\n- rally-tox-cover http://logs.openstack.org/27/653927/3/check/rally-tox-cover/efd9cc4/ : SUCCESS in 3m 43s\n- rally-tox-functional http://logs.openstack.org/27/653927/3/check/rally-tox-functional/f558f4d/ : SUCCESS in 5m 04s\n- rally-tox-functional-py3 http://logs.openstack.org/27/653927/3/check/rally-tox-functional-py3/0d375ae/ : SUCCESS in 5m 03s\n- rally-database-migration http://logs.openstack.org/27/653927/3/check/rally-database-migration/d95e19d/ : SUCCESS in 7m 30s\n- rally-install-ubuntu-xenial http://logs.openstack.org/27/653927/3/check/rally-install-ubuntu-xenial/8986006/ : SUCCESS in 3m 41s\n- rally-install-ubuntu-bionic http://logs.openstack.org/27/653927/3/check/rally-install-ubuntu-bionic/5ccf32c/ : SUCCESS in 3m 20s\n- rally-install-centos-7 http://logs.openstack.org/27/653927/3/check/rally-install-centos-7/7676611/ : SUCCESS in 4m 36s\n- rally-tox-self http://logs.openstack.org/27/653927/3/check/rally-tox-self/c80f4fc/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6afb079f628611e33ca04fde8a9b5f52b47dfab9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-04-20 18:22:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d7bc84b20197b2ea6181b9a6aeb0146afeef63d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 18:22:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"735d9b13c221a02eaa14c6ba83eba78b4a577355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 18:39:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e70640b7304364e9036570a114eb7bb4a09d39ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 18:39:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/653927/3/gate/openstack-tox-docs/a3340e4/html/ : SUCCESS in 3m 13s\n- rally-tox-docs http://logs.openstack.org/27/653927/3/gate/rally-tox-docs/653f68d/ : SUCCESS in 3m 28s\n- rally-tox-pep8 http://logs.openstack.org/27/653927/3/gate/rally-tox-pep8/b091056/ : SUCCESS in 3m 23s\n- rally-tox-py27 http://logs.openstack.org/27/653927/3/gate/rally-tox-py27/a709803/ : SUCCESS in 3m 28s\n- rally-tox-py34 http://logs.openstack.org/27/653927/3/gate/rally-tox-py34/c35e4d1/ : SUCCESS in 5m 09s\n- rally-tox-py35 http://logs.openstack.org/27/653927/3/gate/rally-tox-py35/f91381b/ : SUCCESS in 4m 45s\n- rally-tox-py36 http://logs.openstack.org/27/653927/3/gate/rally-tox-py36/e6bcc86/ : SUCCESS in 3m 05s\n- rally-tox-py37 http://logs.openstack.org/27/653927/3/gate/rally-tox-py37/7ab6d15/ : SUCCESS in 5m 08s\n- rally-tox-cover http://logs.openstack.org/27/653927/3/gate/rally-tox-cover/7823b0a/ : SUCCESS in 4m 03s\n- rally-tox-functional http://logs.openstack.org/27/653927/3/gate/rally-tox-functional/71d8988/ : SUCCESS in 4m 52s\n- rally-tox-functional-py3 http://logs.openstack.org/27/653927/3/gate/rally-tox-functional-py3/9401d39/ : SUCCESS in 5m 54s\n- rally-database-migration http://logs.openstack.org/27/653927/3/gate/rally-database-migration/46e6425/ : SUCCESS in 7m 19s\n- rally-install-ubuntu-xenial http://logs.openstack.org/27/653927/3/gate/rally-install-ubuntu-xenial/bdb7d97/ : SUCCESS in 3m 37s\n- rally-install-ubuntu-bionic http://logs.openstack.org/27/653927/3/gate/rally-install-ubuntu-bionic/0ea5c6a/ : SUCCESS in 3m 33s\n- rally-install-centos-7 http://logs.openstack.org/27/653927/3/gate/rally-install-centos-7/f7fb9c4/ : SUCCESS in 4m 42s\n- rally-tox-self http://logs.openstack.org/27/653927/3/gate/rally-tox-self/9475b7c/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4a1f8549485ab05a0cbb9c289f3339613943344","revisions":{"65083f7b284e07d9710f3d723d18dbfa17675dd6":{"kind":"REWORK","_number":1,"created":"2019-04-19 08:43:57.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/27/653927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/27/653927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/27/653927/1"}}},"commit":{"parents":[{"commit":"5d9336fe29f74900b9f0a704dad356d66f7e6087","subject":"Migrate rally-database-migration to Ubuntu Bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5d9336fe29f74900b9f0a704dad356d66f7e6087"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-19 08:45:20.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-19 08:45:20.000000000","tz":480},"subject":"update upper-constraints.txt for Sphinx","message":"update upper-constraints.txt for Sphinx\n\nSphinx does not support python2.7 in latest version.\n\nChange-Id: Iaa8182c010923eaf052221dcb08b1bd95f7d4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/65083f7b284e07d9710f3d723d18dbfa17675dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/65083f7b284e07d9710f3d723d18dbfa17675dd6"}]},"branch":"refs/heads/master"},"d8a4791afb5470a5528f59877d13c31ba43dc39a":{"kind":"REWORK","_number":2,"created":"2019-04-20 03:53:02.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/27/653927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/27/653927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/27/653927/2"}}},"commit":{"parents":[{"commit":"5d9336fe29f74900b9f0a704dad356d66f7e6087","subject":"Migrate rally-database-migration to Ubuntu Bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5d9336fe29f74900b9f0a704dad356d66f7e6087"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-19 08:45:20.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-20 03:52:52.000000000","tz":0},"subject":"update upper-constraints.txt for Sphinx","message":"update upper-constraints.txt for Sphinx\n\nSphinx does not support python2.7 in latest version.\n\nChange-Id: Iaa8182c010923eaf052221dcb08b1bd95f7d4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d8a4791afb5470a5528f59877d13c31ba43dc39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/d8a4791afb5470a5528f59877d13c31ba43dc39a"}]},"branch":"refs/heads/master"},"c4a1f8549485ab05a0cbb9c289f3339613943344":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-20 03:53:10.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/27/653927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/27/653927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/27/653927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/27/653927/3"}}},"commit":{"parents":[{"commit":"0c667db6859c6b3a54f8b615e574661da9102b0b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/0c667db6859c6b3a54f8b615e574661da9102b0b"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-19 08:45:20.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-20 03:53:10.000000000","tz":0},"subject":"update upper-constraints.txt for Sphinx","message":"update upper-constraints.txt for Sphinx\n\nSphinx does not support python2.7 in latest version.\n\nChange-Id: Iaa8182c010923eaf052221dcb08b1bd95f7d4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c4a1f8549485ab05a0cbb9c289f3339613943344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c4a1f8549485ab05a0cbb9c289f3339613943344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
