)]}'
{"id":"openstack%2Frally~655285","triplet_id":"openstack%2Frally~master~I76d780247c65e2f3002d388635a9922ea391e7f0","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I76d780247c65e2f3002d388635a9922ea391e7f0","subject":"update git.openstack.org to opendev","status":"MERGED","created":"2019-04-24 01:58:36.000000000","updated":"2019-04-27 17:48:26.000000000","submitted":"2019-04-27 17:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655285-1556387306251-b3e99d47","meta_rev_id":"2fe836ad9a7e4e82b7bf8de788438ccb66a6a124","_number":655285,"virtual_id_number":655285,"owner":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-04-27 17:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-04-27 17:30:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-04-27 17:30:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-27 17:30:15.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-04-27 17:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"528d6667053a14bc32c0fdfe82d630c8d5c85320","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-24 01:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6992c9e1299d55685a751137565dda959624742f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 02:17:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/85/655285/1/check/test-release-openstack/d2b912b/ : SUCCESS in 2m 11s\n- openstack-tox-docs http://logs.openstack.org/85/655285/1/check/openstack-tox-docs/48dfcb5/html/ : SUCCESS in 3m 27s\n- rally-tox-docs http://logs.openstack.org/85/655285/1/check/rally-tox-docs/3396019/ : SUCCESS in 3m 16s\n- rally-tox-pep8 http://logs.openstack.org/85/655285/1/check/rally-tox-pep8/ec3082a/ : SUCCESS in 3m 33s\n- rally-tox-py27 http://logs.openstack.org/85/655285/1/check/rally-tox-py27/9fbb17b/ : SUCCESS in 3m 21s\n- rally-tox-py34 http://logs.openstack.org/85/655285/1/check/rally-tox-py34/88f5a66/ : SUCCESS in 5m 56s\n- rally-tox-py35 http://logs.openstack.org/85/655285/1/check/rally-tox-py35/0213c68/ : SUCCESS in 5m 57s\n- rally-tox-py36 http://logs.openstack.org/85/655285/1/check/rally-tox-py36/71154a5/ : SUCCESS in 3m 27s\n- rally-tox-py37 http://logs.openstack.org/85/655285/1/check/rally-tox-py37/d4543f3/ : SUCCESS in 4m 21s\n- rally-tox-cover http://logs.openstack.org/85/655285/1/check/rally-tox-cover/4e923af/ : SUCCESS in 4m 05s\n- rally-tox-functional http://logs.openstack.org/85/655285/1/check/rally-tox-functional/3a7c4a8/ : SUCCESS in 6m 19s\n- rally-tox-functional-py3 http://logs.openstack.org/85/655285/1/check/rally-tox-functional-py3/d9eeb15/ : SUCCESS in 6m 05s\n- rally-database-migration http://logs.openstack.org/85/655285/1/check/rally-database-migration/c7c9c67/ : SUCCESS in 7m 52s\n- rally-install-ubuntu-xenial http://logs.openstack.org/85/655285/1/check/rally-install-ubuntu-xenial/9bb8847/ : SUCCESS in 3m 35s\n- rally-install-ubuntu-bionic http://logs.openstack.org/85/655285/1/check/rally-install-ubuntu-bionic/018b702/ : SUCCESS in 3m 17s\n- rally-install-centos-7 http://logs.openstack.org/85/655285/1/check/rally-install-centos-7/5a29111/ : SUCCESS in 4m 35s\n- rally-tox-self http://logs.openstack.org/85/655285/1/check/rally-tox-self/f9228f9/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4c3bbdd4925eeb4fa148afbd4ce1e6e8cca1189b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-04-25 13:49:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdf4b41dbd7bd24d51508fbec595cc56de258aae","author":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"date":"2019-04-26 00:00:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"37589f120e191f6566a32f168ad19eed5bacc2a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 00:14:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/85/655285/2/check/test-release-openstack/45938e2/ : SUCCESS in 2m 06s\n- openstack-tox-docs http://logs.openstack.org/85/655285/2/check/openstack-tox-docs/e4d2abe/html/ : SUCCESS in 3m 36s\n- rally-tox-docs http://logs.openstack.org/85/655285/2/check/rally-tox-docs/fcae2b5/ : SUCCESS in 3m 07s\n- rally-tox-pep8 http://logs.openstack.org/85/655285/2/check/rally-tox-pep8/ce2018b/ : SUCCESS in 3m 28s\n- rally-tox-py27 http://logs.openstack.org/85/655285/2/check/rally-tox-py27/d9b7cae/ : SUCCESS in 4m 22s\n- rally-tox-py34 http://logs.openstack.org/85/655285/2/check/rally-tox-py34/a5c9812/ : SUCCESS in 4m 41s\n- rally-tox-py35 http://logs.openstack.org/85/655285/2/check/rally-tox-py35/afbb0f9/ : SUCCESS in 4m 55s\n- rally-tox-py36 http://logs.openstack.org/85/655285/2/check/rally-tox-py36/39d5d4c/ : SUCCESS in 3m 33s\n- rally-tox-py37 http://logs.openstack.org/85/655285/2/check/rally-tox-py37/5753b76/ : SUCCESS in 3m 57s\n- rally-tox-cover http://logs.openstack.org/85/655285/2/check/rally-tox-cover/fcd6ee5/ : SUCCESS in 4m 16s\n- rally-tox-functional http://logs.openstack.org/85/655285/2/check/rally-tox-functional/ada6ecc/ : SUCCESS in 5m 09s\n- rally-tox-functional-py3 http://logs.openstack.org/85/655285/2/check/rally-tox-functional-py3/1460443/ : SUCCESS in 5m 18s\n- rally-database-migration http://logs.openstack.org/85/655285/2/check/rally-database-migration/2cfbdab/ : SUCCESS in 7m 26s\n- rally-install-ubuntu-xenial http://logs.openstack.org/85/655285/2/check/rally-install-ubuntu-xenial/6ee74d8/ : SUCCESS in 3m 44s\n- rally-install-ubuntu-bionic http://logs.openstack.org/85/655285/2/check/rally-install-ubuntu-bionic/3beb38c/ : SUCCESS in 3m 24s\n- rally-install-centos-7 http://logs.openstack.org/85/655285/2/check/rally-install-centos-7/3e7cff8/ : SUCCESS in 5m 02s\n- rally-tox-self http://logs.openstack.org/85/655285/2/check/rally-tox-self/54dbabc/ : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"c56c0047b9689a1fc403cef0984fd05e7151052f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-04-27 17:30:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"84ee387852213f8817a98661fdfe05ce5e2b1803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 17:30:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03211612019091cf0c6e34a227503dc9081b7914","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 17:48:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/85/655285/2/gate/test-release-openstack/2fd0231/ : SUCCESS in 1m 43s\n- openstack-tox-docs http://logs.openstack.org/85/655285/2/gate/openstack-tox-docs/ffbe32f/html/ : SUCCESS in 3m 15s\n- rally-tox-docs http://logs.openstack.org/85/655285/2/gate/rally-tox-docs/e837f8a/ : SUCCESS in 2m 57s\n- rally-tox-pep8 http://logs.openstack.org/85/655285/2/gate/rally-tox-pep8/cbf5cd4/ : SUCCESS in 3m 25s\n- rally-tox-py27 http://logs.openstack.org/85/655285/2/gate/rally-tox-py27/91a9a59/ : SUCCESS in 3m 13s\n- rally-tox-py34 http://logs.openstack.org/85/655285/2/gate/rally-tox-py34/38a1eea/ : SUCCESS in 6m 52s\n- rally-tox-py35 http://logs.openstack.org/85/655285/2/gate/rally-tox-py35/84ca95b/ : SUCCESS in 5m 37s\n- rally-tox-py36 http://logs.openstack.org/85/655285/2/gate/rally-tox-py36/c8a9dd9/ : SUCCESS in 3m 20s\n- rally-tox-py37 http://logs.openstack.org/85/655285/2/gate/rally-tox-py37/62110b8/ : SUCCESS in 4m 46s\n- rally-tox-cover http://logs.openstack.org/85/655285/2/gate/rally-tox-cover/803694d/ : SUCCESS in 4m 00s\n- rally-tox-functional http://logs.openstack.org/85/655285/2/gate/rally-tox-functional/6461e01/ : SUCCESS in 4m 56s\n- rally-tox-functional-py3 http://logs.openstack.org/85/655285/2/gate/rally-tox-functional-py3/d5bae83/ : SUCCESS in 4m 58s\n- rally-database-migration http://logs.openstack.org/85/655285/2/gate/rally-database-migration/85ab901/ : SUCCESS in 6m 41s\n- rally-install-ubuntu-xenial http://logs.openstack.org/85/655285/2/gate/rally-install-ubuntu-xenial/613e292/ : SUCCESS in 3m 33s\n- rally-install-ubuntu-bionic http://logs.openstack.org/85/655285/2/gate/rally-install-ubuntu-bionic/fa97b0c/ : SUCCESS in 3m 33s\n- rally-install-centos-7 http://logs.openstack.org/85/655285/2/gate/rally-install-centos-7/a1f73e9/ : SUCCESS in 4m 34s\n- rally-tox-self http://logs.openstack.org/85/655285/2/gate/rally-tox-self/7c028be/ : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2fe836ad9a7e4e82b7bf8de788438ccb66a6a124","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 17:48:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11c0de7e4e411d447b1f04c83da4d32a8a8a1dfb","revisions":{"c82575c9b51e14f1c79801e06cfc0a456cbcfde6":{"kind":"REWORK","_number":1,"created":"2019-04-24 01:58:36.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/85/655285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/85/655285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/85/655285/1"}}},"commit":{"parents":[{"commit":"611d13a788716acadec14cb1c4d07856bce3c5e5","subject":"Merge \"Fix the feature what creating the deployment from sysenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/611d13a788716acadec14cb1c4d07856bce3c5e5"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 01:57:54.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 01:58:30.000000000","tz":480},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I76d780247c65e2f3002d388635a9922ea391e7f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c82575c9b51e14f1c79801e06cfc0a456cbcfde6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c82575c9b51e14f1c79801e06cfc0a456cbcfde6"}]},"branch":"refs/heads/master"},"11c0de7e4e411d447b1f04c83da4d32a8a8a1dfb":{"kind":"REWORK","_number":2,"created":"2019-04-26 00:00:06.000000000","uploader":{"_account_id":30298,"name":"liuwei","email":"liuw147@chinaunicom.cn","username":"ivy123"},"ref":"refs/changes/85/655285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/85/655285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/85/655285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/85/655285/2"}}},"commit":{"parents":[{"commit":"611d13a788716acadec14cb1c4d07856bce3c5e5","subject":"Merge \"Fix the feature what creating the deployment from sysenv\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/611d13a788716acadec14cb1c4d07856bce3c5e5"}]}],"author":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-24 01:57:54.000000000","tz":480},"committer":{"name":"liuwei","email":"liuw147@chinaunicom.cn","date":"2019-04-25 23:59:52.000000000","tz":0},"subject":"update git.openstack.org to opendev","message":"update git.openstack.org to opendev\n\nChange-Id: I76d780247c65e2f3002d388635a9922ea391e7f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/11c0de7e4e411d447b1f04c83da4d32a8a8a1dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/11c0de7e4e411d447b1f04c83da4d32a8a8a1dfb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
