)]}'
{"id":"openstack%2Frally~659197","triplet_id":"openstack%2Frally~master~I61588eaa070b10b780c92d54d1b9439e8fd3f6b6","project":"openstack/rally","branch":"master","topic":"bug/1829030","hashtags":[],"change_id":"I61588eaa070b10b780c92d54d1b9439e8fd3f6b6","subject":"Release 1.5.1 with a fix for os deployment create from env","status":"MERGED","created":"2019-05-15 01:07:00.000000000","updated":"2019-05-16 05:48:07.000000000","submitted":"2019-05-16 05:48:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659197-1557985687968-b2ea1c39","meta_rev_id":"d05cc73b01882ff9c85fb7d537edef81dc5910ef","_number":659197,"virtual_id_number":659197,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-05-16 05:48:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":2,"date":"2019-05-16 05:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":1,"date":"2019-05-16 05:35:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 00:44:19.000000000","updated_by":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"reviewer":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"state":"REVIEWER"},{"updated":"2019-05-16 05:48:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5372e6f66d5f7f7afb2d461e17da676425f48e53","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-15 01:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10f1902f20d13ecf27a60491d8660242b8fde3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 01:20:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/97/659197/1/check/openstack-tox-docs/cb1e3a2/html/ : SUCCESS in 3m 17s\n- rally-tox-docs http://logs.openstack.org/97/659197/1/check/rally-tox-docs/d8cd4d6/ : SUCCESS in 3m 33s\n- rally-tox-pep8 http://logs.openstack.org/97/659197/1/check/rally-tox-pep8/460dabd/ : SUCCESS in 3m 36s\n- rally-tox-py27 http://logs.openstack.org/97/659197/1/check/rally-tox-py27/f077d3d/ : SUCCESS in 3m 01s\n- rally-tox-py34 http://logs.openstack.org/97/659197/1/check/rally-tox-py34/ce07743/ : FAILURE in 4m 38s\n- rally-tox-py35 http://logs.openstack.org/97/659197/1/check/rally-tox-py35/a8fe73d/ : SUCCESS in 5m 58s\n- rally-tox-py36 http://logs.openstack.org/97/659197/1/check/rally-tox-py36/c2e1811/ : SUCCESS in 4m 26s\n- rally-tox-py37 http://logs.openstack.org/97/659197/1/check/rally-tox-py37/fe7b760/ : SUCCESS in 4m 10s\n- rally-tox-cover http://logs.openstack.org/97/659197/1/check/rally-tox-cover/c8e5ea6/ : FAILURE in 3m 51s\n- rally-tox-functional http://logs.openstack.org/97/659197/1/check/rally-tox-functional/f5006ce/ : SUCCESS in 5m 35s\n- rally-tox-functional-py3 http://logs.openstack.org/97/659197/1/check/rally-tox-functional-py3/779916d/ : SUCCESS in 5m 16s\n- rally-database-migration http://logs.openstack.org/97/659197/1/check/rally-database-migration/59e3733/ : SUCCESS in 7m 45s\n- rally-install-ubuntu-xenial http://logs.openstack.org/97/659197/1/check/rally-install-ubuntu-xenial/24b0391/ : SUCCESS in 4m 06s\n- rally-install-ubuntu-bionic http://logs.openstack.org/97/659197/1/check/rally-install-ubuntu-bionic/1239dac/ : SUCCESS in 4m 05s\n- rally-install-centos-7 http://logs.openstack.org/97/659197/1/check/rally-install-centos-7/5d35f95/ : SUCCESS in 4m 46s\n- rally-tox-self http://logs.openstack.org/97/659197/1/check/rally-tox-self/6ebf07f/ : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"48c3a3e7395a3610a10d3739c603579f34ba0839","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-05-16 00:44:19.000000000","message":"Patch Set 1:\n\nlooks it is the same with https://github.com/openstack/rally-openstack/blob/1.4.0/rally_openstack/platforms/existing.py#L138-L140\n\nMy opinion is that we should remove https_cert and https_key from config if both they are empty.","accounts_in_message":[],"_revision_number":1},{"id":"6308827d027d084e1cbfed1a99e8737b38cfbd87","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 04:00:08.000000000","message":"Patch Set 1:\n\n@chen: yes, I copy-pasted it from there.\n\nI\u0027m trying to make this code working with both rally-openstack \u003c\u003d1.4.0 and latest where the proper fix were merged...\n\nif I do not find a solution, maybe poping out both variables can be a solution","accounts_in_message":[],"_revision_number":1},{"id":"2e79d5d077ac24e72faca07922c6f3d4a41302d7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 04:19:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6f0741cac509b3ce053cbe8e4d53df05bafab04","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 04:53:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3c0d8f91e6e71e86474cf767daa8bbfcc56ef57c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:06:16.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/97/659197/3/check/openstack-tox-docs/5bfdb93/html/ : SUCCESS in 3m 15s\n- rally-tox-docs http://logs.openstack.org/97/659197/3/check/rally-tox-docs/8b940ee/ : SUCCESS in 3m 29s\n- rally-tox-pep8 http://logs.openstack.org/97/659197/3/check/rally-tox-pep8/36de17e/ : SUCCESS in 3m 34s\n- rally-tox-py27 http://logs.openstack.org/97/659197/3/check/rally-tox-py27/1b02365/ : SUCCESS in 4m 00s\n- rally-tox-py34 http://logs.openstack.org/97/659197/3/check/rally-tox-py34/3b99014/ : SUCCESS in 6m 48s\n- rally-tox-py35 http://logs.openstack.org/97/659197/3/check/rally-tox-py35/cd11989/ : SUCCESS in 5m 53s\n- rally-tox-py36 http://logs.openstack.org/97/659197/3/check/rally-tox-py36/5aab868/ : SUCCESS in 3m 17s\n- rally-tox-py37 http://logs.openstack.org/97/659197/3/check/rally-tox-py37/7e079d2/ : SUCCESS in 5m 25s\n- rally-tox-cover http://logs.openstack.org/97/659197/3/check/rally-tox-cover/c2b1f37/ : FAILURE in 4m 03s\n- rally-tox-functional http://logs.openstack.org/97/659197/3/check/rally-tox-functional/6eccf72/ : SUCCESS in 5m 05s\n- rally-tox-functional-py3 http://logs.openstack.org/97/659197/3/check/rally-tox-functional-py3/098460e/ : SUCCESS in 5m 40s\n- rally-database-migration http://logs.openstack.org/97/659197/3/check/rally-database-migration/3cd19de/ : SUCCESS in 7m 34s\n- rally-install-ubuntu-xenial http://logs.openstack.org/97/659197/3/check/rally-install-ubuntu-xenial/1340628/ : SUCCESS in 3m 52s\n- rally-install-ubuntu-bionic http://logs.openstack.org/97/659197/3/check/rally-install-ubuntu-bionic/cbb54a0/ : SUCCESS in 3m 43s\n- rally-install-centos-7 http://logs.openstack.org/97/659197/3/check/rally-install-centos-7/4935d8a/ : SUCCESS in 5m 14s\n- rally-tox-self http://logs.openstack.org/97/659197/3/check/rally-tox-self/83041aa/ : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":3},{"id":"b5d432bdf44b1a6b5c0a142fef812933fae16aa8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 05:24:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d33aaeed84b07db113b332fab6ab7f710c449077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:33:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659197/4/check/openstack-tox-docs/a3780f3/html/ : SUCCESS in 2m 52s\n- rally-tox-docs http://logs.openstack.org/97/659197/4/check/rally-tox-docs/3fa47a7/ : SUCCESS in 3m 25s\n- rally-tox-pep8 http://logs.openstack.org/97/659197/4/check/rally-tox-pep8/3d249a8/ : SUCCESS in 4m 44s\n- rally-tox-py27 http://logs.openstack.org/97/659197/4/check/rally-tox-py27/7edf7f9/ : SUCCESS in 3m 16s\n- rally-tox-py34 http://logs.openstack.org/97/659197/4/check/rally-tox-py34/00a328e/ : SUCCESS in 5m 34s\n- rally-tox-py35 http://logs.openstack.org/97/659197/4/check/rally-tox-py35/36e7c2b/ : SUCCESS in 6m 22s\n- rally-tox-py36 http://logs.openstack.org/97/659197/4/check/rally-tox-py36/cd036dd/ : SUCCESS in 2m 44s\n- rally-tox-py37 http://logs.openstack.org/97/659197/4/check/rally-tox-py37/120b690/ : SUCCESS in 4m 03s\n- rally-tox-cover http://logs.openstack.org/97/659197/4/check/rally-tox-cover/d708d1a/ : SUCCESS in 4m 02s\n- rally-tox-functional http://logs.openstack.org/97/659197/4/check/rally-tox-functional/3d9d41c/ : SUCCESS in 5m 28s\n- rally-tox-functional-py3 http://logs.openstack.org/97/659197/4/check/rally-tox-functional-py3/e499c0e/ : SUCCESS in 5m 12s\n- rally-database-migration http://logs.openstack.org/97/659197/4/check/rally-database-migration/d1b5966/ : SUCCESS in 7m 25s\n- rally-install-ubuntu-xenial http://logs.openstack.org/97/659197/4/check/rally-install-ubuntu-xenial/6ec2205/ : SUCCESS in 4m 00s\n- rally-install-ubuntu-bionic http://logs.openstack.org/97/659197/4/check/rally-install-ubuntu-bionic/027658f/ : SUCCESS in 4m 09s\n- rally-install-centos-7 http://logs.openstack.org/97/659197/4/check/rally-install-centos-7/6992ed8/ : SUCCESS in 4m 30s\n- rally-tox-self http://logs.openstack.org/97/659197/4/check/rally-tox-self/f224d4a/ : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":4},{"id":"91b325557cdb9bd42b1e8935ba337d6300a98f04","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-05-16 05:35:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6e4ebddfac481efcece22598cd85b370a3b84f6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:36:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0e160add4b3d23feb5c3428519bacdb33520f202","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:48:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ef35af88d7f780006361e2f6af5ce992e4981f60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:48:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659197/4/gate/openstack-tox-docs/f2eb07c/html/ : SUCCESS in 3m 10s\n- rally-tox-docs http://logs.openstack.org/97/659197/4/gate/rally-tox-docs/e5326ed/ : SUCCESS in 3m 17s\n- rally-tox-pep8 http://logs.openstack.org/97/659197/4/gate/rally-tox-pep8/5ad1a7e/ : SUCCESS in 3m 31s\n- rally-tox-py27 http://logs.openstack.org/97/659197/4/gate/rally-tox-py27/f7d9084/ : SUCCESS in 3m 21s\n- rally-tox-py34 http://logs.openstack.org/97/659197/4/gate/rally-tox-py34/150fb1c/ : SUCCESS in 6m 29s\n- rally-tox-py35 http://logs.openstack.org/97/659197/4/gate/rally-tox-py35/61722d5/ : SUCCESS in 6m 12s\n- rally-tox-py36 http://logs.openstack.org/97/659197/4/gate/rally-tox-py36/a09d89d/ : SUCCESS in 3m 24s\n- rally-tox-py37 http://logs.openstack.org/97/659197/4/gate/rally-tox-py37/32ad360/ : SUCCESS in 4m 34s\n- rally-tox-cover http://logs.openstack.org/97/659197/4/gate/rally-tox-cover/b03306a/ : SUCCESS in 3m 46s\n- rally-tox-functional http://logs.openstack.org/97/659197/4/gate/rally-tox-functional/3d4f92a/ : SUCCESS in 4m 40s\n- rally-tox-functional-py3 http://logs.openstack.org/97/659197/4/gate/rally-tox-functional-py3/786cfb8/ : SUCCESS in 5m 21s\n- rally-database-migration http://logs.openstack.org/97/659197/4/gate/rally-database-migration/36080d3/ : SUCCESS in 7m 25s\n- rally-install-ubuntu-xenial http://logs.openstack.org/97/659197/4/gate/rally-install-ubuntu-xenial/3cbec9d/ : SUCCESS in 3m 37s\n- rally-install-ubuntu-bionic http://logs.openstack.org/97/659197/4/gate/rally-install-ubuntu-bionic/9c0c9ce/ : SUCCESS in 3m 14s\n- rally-install-centos-7 http://logs.openstack.org/97/659197/4/gate/rally-install-centos-7/135bdd3/ : SUCCESS in 5m 56s\n- rally-tox-self http://logs.openstack.org/97/659197/4/gate/rally-tox-self/62a336d/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"14a8b0f931424d42543c100c9deb409eaded86d3","revisions":{"6d968a6fd5115fb611fcf754f4246fe71b3b2518":{"kind":"REWORK","_number":1,"created":"2019-05-15 01:07:00.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/659197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/659197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/659197/1"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:06:57.000000000","tz":-420},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:06:57.000000000","tz":-420},"subject":"[deployment] Fix --from-env option for openstack","message":"[deployment] Fix --from-env option for openstack\n\nChange-Id: I61588eaa070b10b780c92d54d1b9439e8fd3f6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6d968a6fd5115fb611fcf754f4246fe71b3b2518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6d968a6fd5115fb611fcf754f4246fe71b3b2518"}]},"branch":"refs/heads/master"},"163490d8a7b3fde3ddd93cc753a108a1512ab2e4":{"kind":"REWORK","_number":2,"created":"2019-05-16 04:19:23.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/659197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/659197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/659197/2"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:06:57.000000000","tz":-420},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 04:11:07.000000000","tz":-420},"subject":"[deployment] Fix --from-env option for openstack","message":"[deployment] Fix --from-env option for openstack\n\nCloses-Bug: #1829030\nChange-Id: I61588eaa070b10b780c92d54d1b9439e8fd3f6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/163490d8a7b3fde3ddd93cc753a108a1512ab2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/163490d8a7b3fde3ddd93cc753a108a1512ab2e4"}]},"branch":"refs/heads/master"},"85f33eb45dd68c7362f4ca60f3d99d669e54ff6e":{"kind":"REWORK","_number":3,"created":"2019-05-16 04:53:56.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/659197/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/659197/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/659197/3"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:06:57.000000000","tz":-420},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 04:53:42.000000000","tz":-420},"subject":"Release 1.5.1 with a fix for os deployment create from env","message":"Release 1.5.1 with a fix for os deployment create from env\n\n**rally deployment create* --fromenv* creates wrong spec for\nrally-openstack\u003c\u003d1.4.0\n\nCloses-Bug: #1829030\nChange-Id: I61588eaa070b10b780c92d54d1b9439e8fd3f6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/85f33eb45dd68c7362f4ca60f3d99d669e54ff6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/85f33eb45dd68c7362f4ca60f3d99d669e54ff6e"}]},"branch":"refs/heads/master"},"14a8b0f931424d42543c100c9deb409eaded86d3":{"kind":"REWORK","_number":4,"created":"2019-05-16 05:24:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/97/659197/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/97/659197/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/97/659197/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/97/659197/4"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-15 01:06:57.000000000","tz":-420},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-05-16 05:22:46.000000000","tz":-420},"subject":"Release 1.5.1 with a fix for os deployment create from env","message":"Release 1.5.1 with a fix for os deployment create from env\n\n**rally deployment create* --fromenv* creates wrong spec for\nrally-openstack\u003c\u003d1.4.0\n\nCloses-Bug: #1829030\nChange-Id: I61588eaa070b10b780c92d54d1b9439e8fd3f6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14a8b0f931424d42543c100c9deb409eaded86d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14a8b0f931424d42543c100c9deb409eaded86d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
