)]}'
{"id":"openstack%2Frally~661342","triplet_id":"openstack%2Frally~master~I10d7477cd3321683899d455bf4370ec8ef8a4d05","project":"openstack/rally","branch":"master","topic":"regex-rebase","hashtags":[],"change_id":"I10d7477cd3321683899d455bf4370ec8ef8a4d05","subject":"Allow regular expressions in skip-list","status":"MERGED","created":"2019-05-24 15:49:07.000000000","updated":"2019-06-04 20:50:08.000000000","submitted":"2019-06-03 14:30:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661342-1559572223529-fc91e94c","meta_rev_id":"27cbdd1c4a716c213c9b0b443bddf98c22d7936a","_number":661342,"virtual_id_number":661342,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-03 14:30:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-06-03 13:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-06-03 13:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-03 13:59:21.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-06-03 14:30:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff0dd9ee0010b9cd043a77663db7c8884ead22e9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-24 15:49:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"911c2c7c594553b554f59f45fe7d8177fd4d70a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:14:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/42/661342/1/check/openstack-tox-docs/75a3f13/html/ : SUCCESS in 3m 17s\n- rally-tox-docs http://logs.openstack.org/42/661342/1/check/rally-tox-docs/6cf1939/ : SUCCESS in 3m 37s\n- rally-tox-pep8 http://logs.openstack.org/42/661342/1/check/rally-tox-pep8/0b73dbb/ : FAILURE in 3m 47s\n- rally-tox-py27 http://logs.openstack.org/42/661342/1/check/rally-tox-py27/b83c751/ : FAILURE in 3m 58s\n- rally-tox-py34 http://logs.openstack.org/42/661342/1/check/rally-tox-py34/3daa021/ : FAILURE in 6m 19s\n- rally-tox-py35 http://logs.openstack.org/42/661342/1/check/rally-tox-py35/761ff12/ : FAILURE in 6m 33s\n- rally-tox-py36 http://logs.openstack.org/42/661342/1/check/rally-tox-py36/2657813/ : FAILURE in 3m 18s\n- rally-tox-py37 http://logs.openstack.org/42/661342/1/check/rally-tox-py37/66eb1e4/ : FAILURE in 4m 16s\n- rally-tox-cover http://logs.openstack.org/42/661342/1/check/rally-tox-cover/578772f/ : SUCCESS in 3m 50s\n- rally-tox-functional http://logs.openstack.org/42/661342/1/check/rally-tox-functional/82432d3/ : SUCCESS in 5m 04s\n- rally-tox-functional-py3 http://logs.openstack.org/42/661342/1/check/rally-tox-functional-py3/bc4b07e/ : SUCCESS in 5m 46s\n- rally-database-migration http://logs.openstack.org/42/661342/1/check/rally-database-migration/d9eb0fe/ : SUCCESS in 7m 32s\n- rally-install-ubuntu-xenial http://logs.openstack.org/42/661342/1/check/rally-install-ubuntu-xenial/f6398c1/ : SUCCESS in 3m 24s\n- rally-install-ubuntu-bionic http://logs.openstack.org/42/661342/1/check/rally-install-ubuntu-bionic/db8698c/ : SUCCESS in 3m 37s\n- rally-install-centos-7 http://logs.openstack.org/42/661342/1/check/rally-install-centos-7/8b762d4/ : SUCCESS in 4m 47s\n- rally-tox-self http://logs.openstack.org/42/661342/1/check/rally-tox-self/75920c8/ : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d353e2513047d30f5b54c968c47d6dc1a32766bf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-29 09:35:02.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"038d35f3b93d4bbdeed79235cde458c06fac325f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-05-29 15:04:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af316a791de8a1f5ce6df02388d4921c3c89c12a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 15:33:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/661342/2/check/openstack-tox-docs/fc0e9b3/html/ : SUCCESS in 3m 28s\n- rally-tox-docs http://logs.openstack.org/42/661342/2/check/rally-tox-docs/9491cb1/ : SUCCESS in 3m 57s\n- rally-tox-pep8 http://logs.openstack.org/42/661342/2/check/rally-tox-pep8/47931bf/ : SUCCESS in 3m 48s\n- rally-tox-py27 http://logs.openstack.org/42/661342/2/check/rally-tox-py27/ddc7efc/ : SUCCESS in 3m 27s\n- rally-tox-py34 http://logs.openstack.org/42/661342/2/check/rally-tox-py34/9abb7a7/ : SUCCESS in 6m 49s\n- rally-tox-py35 http://logs.openstack.org/42/661342/2/check/rally-tox-py35/c092223/ : SUCCESS in 6m 38s\n- rally-tox-py36 http://logs.openstack.org/42/661342/2/check/rally-tox-py36/878b0d1/ : SUCCESS in 4m 32s\n- rally-tox-py37 http://logs.openstack.org/42/661342/2/check/rally-tox-py37/5bbe644/ : SUCCESS in 4m 26s\n- rally-tox-cover http://logs.openstack.org/42/661342/2/check/rally-tox-cover/8ff9d80/ : SUCCESS in 4m 33s\n- rally-tox-functional http://logs.openstack.org/42/661342/2/check/rally-tox-functional/8f64a74/ : SUCCESS in 5m 32s\n- rally-tox-functional-py3 http://logs.openstack.org/42/661342/2/check/rally-tox-functional-py3/2fba0df/ : SUCCESS in 5m 20s\n- rally-database-migration http://logs.openstack.org/42/661342/2/check/rally-database-migration/a75dcdf/ : SUCCESS in 7m 36s\n- rally-install-ubuntu-xenial http://logs.openstack.org/42/661342/2/check/rally-install-ubuntu-xenial/569dbe1/ : SUCCESS in 3m 57s\n- rally-install-ubuntu-bionic http://logs.openstack.org/42/661342/2/check/rally-install-ubuntu-bionic/f227771/ : SUCCESS in 4m 06s\n- rally-install-centos-7 http://logs.openstack.org/42/661342/2/check/rally-install-centos-7/1cd731f/ : SUCCESS in 5m 57s\n- rally-tox-self http://logs.openstack.org/42/661342/2/check/rally-tox-self/e41dc5e/ : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":2},{"id":"48a497f952a25908577183e90a5f5df3638bd917","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-03 13:59:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks for the change. It is a really great feature.","accounts_in_message":[],"_revision_number":2},{"id":"4e6f3f9e697ffc627d8a6c4e260791c3cfe75fd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 13:59:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e378adaabbaa1f1e496136026984bbf458026c9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 14:30:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"87837e51e9e3889aba6c5644f36e97bb41b6b054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 14:30:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/42/661342/2/gate/openstack-tox-docs/010ff44/html/ : SUCCESS in 4m 06s\n- rally-tox-docs http://logs.openstack.org/42/661342/2/gate/rally-tox-docs/1ecfbd2/ : SUCCESS in 3m 48s\n- rally-tox-pep8 http://logs.openstack.org/42/661342/2/gate/rally-tox-pep8/4d3f9a6/ : SUCCESS in 3m 51s\n- rally-tox-py27 http://logs.openstack.org/42/661342/2/gate/rally-tox-py27/cc85eab/ : SUCCESS in 4m 01s\n- rally-tox-py34 http://logs.openstack.org/42/661342/2/gate/rally-tox-py34/255ea45/ : SUCCESS in 7m 10s\n- rally-tox-py35 http://logs.openstack.org/42/661342/2/gate/rally-tox-py35/c0c17d4/ : SUCCESS in 7m 18s\n- rally-tox-py36 http://logs.openstack.org/42/661342/2/gate/rally-tox-py36/2f02e86/ : SUCCESS in 3m 46s\n- rally-tox-py37 http://logs.openstack.org/42/661342/2/gate/rally-tox-py37/3714d03/ : SUCCESS in 5m 23s\n- rally-tox-cover http://logs.openstack.org/42/661342/2/gate/rally-tox-cover/72da31e/ : SUCCESS in 4m 04s\n- rally-tox-functional http://logs.openstack.org/42/661342/2/gate/rally-tox-functional/48c0e61/ : SUCCESS in 6m 40s\n- rally-tox-functional-py3 http://logs.openstack.org/42/661342/2/gate/rally-tox-functional-py3/aeb1928/ : SUCCESS in 10m 01s\n- rally-database-migration http://logs.openstack.org/42/661342/2/gate/rally-database-migration/443b03b/ : SUCCESS in 8m 12s\n- rally-install-ubuntu-xenial http://logs.openstack.org/42/661342/2/gate/rally-install-ubuntu-xenial/e317bc2/ : SUCCESS in 4m 14s\n- rally-install-ubuntu-bionic http://logs.openstack.org/42/661342/2/gate/rally-install-ubuntu-bionic/6e1f207/ : SUCCESS in 4m 15s\n- rally-install-centos-7 http://logs.openstack.org/42/661342/2/gate/rally-install-centos-7/ec78e76/ : SUCCESS in 6m 02s\n- rally-tox-self http://logs.openstack.org/42/661342/2/gate/rally-tox-self/231d163/ : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8f0d5ec8614ec2b1b08fdf5da4b193b8668c2e71","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-06-04 09:20:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6c4e2ab5ade41d36e73884d16711bb7795eda85","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-04 20:50:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f99f4422497b19d63cf7e69be78760493c1774bf","revisions":{"a2e26196e31b28c6fcc93aeefa7f72f809900bb1":{"kind":"REWORK","_number":1,"created":"2019-05-24 15:49:07.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/661342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/42/661342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/42/661342/1"}}},"commit":{"parents":[{"commit":"14a8b0f931424d42543c100c9deb409eaded86d3","subject":"Release 1.5.1 with a fix for os deployment create from env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14a8b0f931424d42543c100c9deb409eaded86d3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-05-24 15:03:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-05-24 15:08:52.000000000","tz":60},"subject":"Allow regular expressions in skip-list","message":"Allow regular expressions in skip-list\n\nThis allows you to more easily skip tests belonging to particular class,\nas you don\u0027t have to list them individually.\n\nBackwards compatability is maintained by treating invalid regexes\nas literal strings.\n\nChange-Id: I10d7477cd3321683899d455bf4370ec8ef8a4d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a2e26196e31b28c6fcc93aeefa7f72f809900bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a2e26196e31b28c6fcc93aeefa7f72f809900bb1"}]},"branch":"refs/heads/master"},"f99f4422497b19d63cf7e69be78760493c1774bf":{"kind":"REWORK","_number":2,"created":"2019-05-29 15:04:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/42/661342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/42/661342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/42/661342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/42/661342/2"}}},"commit":{"parents":[{"commit":"14a8b0f931424d42543c100c9deb409eaded86d3","subject":"Release 1.5.1 with a fix for os deployment create from env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/14a8b0f931424d42543c100c9deb409eaded86d3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-05-24 15:03:11.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-05-29 13:10:29.000000000","tz":60},"subject":"Allow regular expressions in skip-list","message":"Allow regular expressions in skip-list\n\nThis allows you to more easily skip tests belonging to particular class,\nas you don\u0027t have to list them individually.\n\nBackwards compatability is maintained by treating invalid regexes\nas literal strings.\n\nChange-Id: I10d7477cd3321683899d455bf4370ec8ef8a4d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f99f4422497b19d63cf7e69be78760493c1774bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f99f4422497b19d63cf7e69be78760493c1774bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
