)]}'
{"id":"openstack%2Frally~662856","triplet_id":"openstack%2Frally~master~I9a30b7d5a674e2bd444e3a286d672d6c589e6712","project":"openstack/rally","branch":"master","topic":"pkg_manager_order","hashtags":[],"change_id":"I9a30b7d5a674e2bd444e3a286d672d6c589e6712","subject":"install_rally.sh: moved zypper pkg manager up to handle case when yum is installed on SUSE to fulfill dependencies.","status":"MERGED","created":"2019-06-03 20:58:36.000000000","updated":"2019-06-17 12:27:57.000000000","submitted":"2019-06-17 12:27:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662856-1560774477954-ad3aa802","meta_rev_id":"152f6e139f1c81807ff1b9bc9bd82b309c54a3ff","_number":662856,"virtual_id_number":662856,"owner":{"_account_id":30084,"name":"Nefi Munoz","email":"nefi.munoz@gmail.com","username":"nflondo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-17 11:41:58.000000000","_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-17 12:27:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":2,"date":"2019-06-14 15:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":1,"date":"2019-06-14 15:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 15:59:58.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-06-17 11:41:58.000000000","updated_by":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"reviewer":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"state":"REVIEWER"},{"updated":"2019-06-17 12:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb4f5c3cf42fc8a0940c1350c708fd48059cfdf6","author":{"_account_id":30084,"name":"Nefi Munoz","email":"nefi.munoz@gmail.com","username":"nflondo"},"date":"2019-06-03 20:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48e349b1be7852a0be27f64088af6c37015865e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/662856/1/check/openstack-tox-docs/a8cc3b2/html/ : SUCCESS in 3m 43s\n- rally-tox-docs http://logs.openstack.org/56/662856/1/check/rally-tox-docs/e4c70e0/ : SUCCESS in 3m 32s\n- rally-tox-pep8 http://logs.openstack.org/56/662856/1/check/rally-tox-pep8/a886060/ : SUCCESS in 4m 13s\n- rally-tox-py27 http://logs.openstack.org/56/662856/1/check/rally-tox-py27/a8ac079/ : SUCCESS in 3m 21s\n- rally-tox-py34 http://logs.openstack.org/56/662856/1/check/rally-tox-py34/740a20e/ : SUCCESS in 6m 08s\n- rally-tox-py35 http://logs.openstack.org/56/662856/1/check/rally-tox-py35/56dad48/ : SUCCESS in 6m 13s\n- rally-tox-py36 http://logs.openstack.org/56/662856/1/check/rally-tox-py36/3fc8b5f/ : SUCCESS in 3m 11s\n- rally-tox-py37 http://logs.openstack.org/56/662856/1/check/rally-tox-py37/ecf0ce8/ : SUCCESS in 4m 32s\n- rally-tox-cover http://logs.openstack.org/56/662856/1/check/rally-tox-cover/81742bc/ : SUCCESS in 4m 24s\n- rally-tox-functional http://logs.openstack.org/56/662856/1/check/rally-tox-functional/e5bee6d/ : SUCCESS in 5m 15s\n- rally-tox-functional-py3 http://logs.openstack.org/56/662856/1/check/rally-tox-functional-py3/1d388d3/ : SUCCESS in 5m 47s\n- rally-database-migration http://logs.openstack.org/56/662856/1/check/rally-database-migration/20d3ce3/ : SUCCESS in 7m 25s\n- rally-install-ubuntu-xenial http://logs.openstack.org/56/662856/1/check/rally-install-ubuntu-xenial/2b0d7a6/ : SUCCESS in 4m 13s\n- rally-install-ubuntu-bionic http://logs.openstack.org/56/662856/1/check/rally-install-ubuntu-bionic/e1660fd/ : SUCCESS in 3m 56s\n- rally-install-centos-7 http://logs.openstack.org/56/662856/1/check/rally-install-centos-7/67ee6d4/ : SUCCESS in 4m 23s\n- rally-tox-self http://logs.openstack.org/56/662856/1/check/rally-tox-self/1a88e86/ : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":1},{"id":"160ce59b7c83d09c6a473d2711c7fac1713503c4","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-14 15:59:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6060574a2059e3298346489c414b92b7bfa4d513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:00:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"934792384400a01ef150b025195178e41268a45b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:32:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/56/662856/1/gate/openstack-tox-docs/18339c1/html/ : SUCCESS in 3m 29s\n- rally-tox-docs http://logs.openstack.org/56/662856/1/gate/rally-tox-docs/9236c06/ : SUCCESS in 4m 10s\n- rally-tox-pep8 http://logs.openstack.org/56/662856/1/gate/rally-tox-pep8/997bbe1/ : SUCCESS in 4m 31s\n- rally-tox-py27 http://logs.openstack.org/56/662856/1/gate/rally-tox-py27/b59e186/ : SUCCESS in 3m 52s\n- rally-tox-py34 http://logs.openstack.org/56/662856/1/gate/rally-tox-py34/38d034f/ : SUCCESS in 6m 05s\n- rally-tox-py35 http://logs.openstack.org/56/662856/1/gate/rally-tox-py35/4250a32/ : SUCCESS in 6m 37s\n- rally-tox-py36 http://logs.openstack.org/56/662856/1/gate/rally-tox-py36/2e50db9/ : SUCCESS in 4m 29s\n- rally-tox-py37 http://logs.openstack.org/56/662856/1/gate/rally-tox-py37/b67ff1a/ : SUCCESS in 5m 06s\n- rally-tox-cover http://logs.openstack.org/56/662856/1/gate/rally-tox-cover/0da947c/ : SUCCESS in 5m 54s\n- rally-tox-functional http://logs.openstack.org/56/662856/1/gate/rally-tox-functional/f7cc0b5/ : SUCCESS in 7m 29s\n- rally-tox-functional-py3 http://logs.openstack.org/56/662856/1/gate/rally-tox-functional-py3/16e6433/ : SUCCESS in 5m 45s\n- rally-database-migration http://logs.openstack.org/56/662856/1/gate/rally-database-migration/d07b930/ : SUCCESS in 7m 44s\n- rally-install-ubuntu-xenial http://logs.openstack.org/56/662856/1/gate/rally-install-ubuntu-xenial/8c86a9a/ : SUCCESS in 3m 29s\n- rally-install-ubuntu-bionic http://logs.openstack.org/56/662856/1/gate/rally-install-ubuntu-bionic/3f539e0/ : TIMED_OUT in 30m 42s\n- rally-install-centos-7 http://logs.openstack.org/56/662856/1/gate/rally-install-centos-7/efe7e5a/ : SUCCESS in 5m 15s\n- rally-tox-self http://logs.openstack.org/56/662856/1/gate/rally-tox-self/5ccb671/ : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"40c95a18d562b4e4c903ef0ed596d8715a545273","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-17 11:41:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9e8f87cec5f52262eafaa0116933c2be12e747b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/662856/1/check/openstack-tox-docs/d4779ca/html/ : SUCCESS in 4m 19s\n- rally-tox-docs http://logs.openstack.org/56/662856/1/check/rally-tox-docs/25ecb50/ : SUCCESS in 3m 29s\n- rally-tox-pep8 http://logs.openstack.org/56/662856/1/check/rally-tox-pep8/fb46ae8/ : SUCCESS in 4m 06s\n- rally-tox-py27 http://logs.openstack.org/56/662856/1/check/rally-tox-py27/2446aa5/ : SUCCESS in 4m 06s\n- rally-tox-py34 http://logs.openstack.org/56/662856/1/check/rally-tox-py34/bd64b23/ : SUCCESS in 4m 45s\n- rally-tox-py35 http://logs.openstack.org/56/662856/1/check/rally-tox-py35/57f0714/ : SUCCESS in 4m 55s\n- rally-tox-py36 http://logs.openstack.org/56/662856/1/check/rally-tox-py36/3a9dce8/ : SUCCESS in 3m 39s\n- rally-tox-py37 http://logs.openstack.org/56/662856/1/check/rally-tox-py37/5965204/ : SUCCESS in 4m 44s\n- rally-tox-cover http://logs.openstack.org/56/662856/1/check/rally-tox-cover/1bc5a98/ : SUCCESS in 4m 42s\n- rally-tox-functional http://logs.openstack.org/56/662856/1/check/rally-tox-functional/9d12cdf/ : SUCCESS in 5m 30s\n- rally-tox-functional-py3 http://logs.openstack.org/56/662856/1/check/rally-tox-functional-py3/249fa02/ : SUCCESS in 5m 46s\n- rally-database-migration http://logs.openstack.org/56/662856/1/check/rally-database-migration/f3cf180/ : SUCCESS in 7m 35s\n- rally-install-ubuntu-xenial http://logs.openstack.org/56/662856/1/check/rally-install-ubuntu-xenial/7a00214/ : SUCCESS in 3m 46s\n- rally-install-ubuntu-bionic http://logs.openstack.org/56/662856/1/check/rally-install-ubuntu-bionic/64f1625/ : SUCCESS in 3m 41s\n- rally-install-centos-7 http://logs.openstack.org/56/662856/1/check/rally-install-centos-7/fa36711/ : SUCCESS in 4m 46s\n- rally-tox-self http://logs.openstack.org/56/662856/1/check/rally-tox-self/20de0f9/ : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"40cbcc220ba80eeb34ec01b64b72124a81a1aae2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 11:57:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f13f6679f482aca9c25f3cf8542a36d16013259","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:27:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/662856/1/gate/openstack-tox-docs/863450f/html/ : SUCCESS in 3m 14s\n- rally-tox-docs http://logs.openstack.org/56/662856/1/gate/rally-tox-docs/e30439d/ : SUCCESS in 4m 12s\n- rally-tox-pep8 http://logs.openstack.org/56/662856/1/gate/rally-tox-pep8/304370b/ : SUCCESS in 4m 26s\n- rally-tox-py27 http://logs.openstack.org/56/662856/1/gate/rally-tox-py27/b71ee32/ : SUCCESS in 3m 55s\n- rally-tox-py34 http://logs.openstack.org/56/662856/1/gate/rally-tox-py34/9cac167/ : SUCCESS in 4m 58s\n- rally-tox-py35 http://logs.openstack.org/56/662856/1/gate/rally-tox-py35/a12ec37/ : SUCCESS in 5m 10s\n- rally-tox-py36 http://logs.openstack.org/56/662856/1/gate/rally-tox-py36/8c0c1a7/ : SUCCESS in 3m 44s\n- rally-tox-py37 http://logs.openstack.org/56/662856/1/gate/rally-tox-py37/e56a74e/ : SUCCESS in 4m 10s\n- rally-tox-cover http://logs.openstack.org/56/662856/1/gate/rally-tox-cover/d1f5554/ : SUCCESS in 4m 26s\n- rally-tox-functional http://logs.openstack.org/56/662856/1/gate/rally-tox-functional/52186bb/ : SUCCESS in 5m 19s\n- rally-tox-functional-py3 http://logs.openstack.org/56/662856/1/gate/rally-tox-functional-py3/b8199d1/ : SUCCESS in 6m 07s\n- rally-database-migration http://logs.openstack.org/56/662856/1/gate/rally-database-migration/459148e/ : SUCCESS in 7m 53s\n- rally-install-ubuntu-xenial http://logs.openstack.org/56/662856/1/gate/rally-install-ubuntu-xenial/b12f919/ : SUCCESS in 4m 12s\n- rally-install-ubuntu-bionic http://logs.openstack.org/56/662856/1/gate/rally-install-ubuntu-bionic/4974ca0/ : SUCCESS in 4m 41s\n- rally-install-centos-7 http://logs.openstack.org/56/662856/1/gate/rally-install-centos-7/d7bca8a/ : SUCCESS in 5m 19s\n- rally-tox-self http://logs.openstack.org/56/662856/1/gate/rally-tox-self/c90faca/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"edeae07c5bfbec94f7f875d7a9e3e3ea35ea0d73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 12:27:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcdcfc7900aad26635df793c90b36d3f2e6bf386","revisions":{"bcdcfc7900aad26635df793c90b36d3f2e6bf386":{"kind":"REWORK","_number":1,"created":"2019-06-03 20:58:36.000000000","uploader":{"_account_id":30084,"name":"Nefi Munoz","email":"nefi.munoz@gmail.com","username":"nflondo"},"ref":"refs/changes/56/662856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/56/662856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/56/662856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/56/662856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/56/662856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/56/662856/1"}}},"commit":{"parents":[{"commit":"075843292124cdcbfcd71bf99b200da276d8c07d","subject":"Merge \"Allow regular expressions in skip-list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/075843292124cdcbfcd71bf99b200da276d8c07d"}]}],"author":{"name":"Nefi Munoz","email":"nmunoz@suse.com","date":"2019-06-03 20:43:11.000000000","tz":-360},"committer":{"name":"Nefi Munoz","email":"nmunoz@suse.com","date":"2019-06-03 20:43:11.000000000","tz":-360},"subject":"install_rally.sh: moved zypper pkg manager up to handle case when yum is installed on SUSE to fulfill dependencies.","message":"install_rally.sh: moved zypper pkg manager up to handle case when\nyum is installed on SUSE to fulfill dependencies.\n\nChange-Id: I9a30b7d5a674e2bd444e3a286d672d6c589e6712\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bcdcfc7900aad26635df793c90b36d3f2e6bf386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bcdcfc7900aad26635df793c90b36d3f2e6bf386"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
