)]}'
{"id":"openstack%2Frally~665918","triplet_id":"openstack%2Frally~master~Id1955160813c9be87dffc7e38b8d3476cec04d60","project":"openstack/rally","branch":"master","topic":"dddd","hashtags":[],"change_id":"Id1955160813c9be87dffc7e38b8d3476cec04d60","subject":"Update SQLAlchemy in  upper-constraints.txt","status":"MERGED","created":"2019-06-18 09:04:22.000000000","updated":"2019-06-19 13:41:00.000000000","submitted":"2019-06-19 13:41:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665918-1560951660511-eb0806c3","meta_rev_id":"675c1cd2249ded1ed10650bb72779ba7a30a0a8d","_number":665918,"virtual_id_number":665918,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-19 13:41:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-06-19 13:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-06-19 13:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 13:19:50.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-06-19 13:41:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"41952764b6967e0485bca01a0e29b223cafa64cf","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-18 09:04:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab93420d320a6a8e1703a085aef46033f4801bda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:28:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/1/check/openstack-tox-docs/6dac1b5/html/ : SUCCESS in 3m 15s\n- rally-tox-docs http://logs.openstack.org/18/665918/1/check/rally-tox-docs/d044154/ : SUCCESS in 3m 40s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/1/check/rally-tox-pep8/fb5772f/ : SUCCESS in 4m 00s\n- rally-tox-py27 http://logs.openstack.org/18/665918/1/check/rally-tox-py27/eb00d5d/ : FAILURE in 4m 06s\n- rally-tox-py34 http://logs.openstack.org/18/665918/1/check/rally-tox-py34/03e09bf/ : FAILURE in 4m 42s\n- rally-tox-py35 http://logs.openstack.org/18/665918/1/check/rally-tox-py35/cec04a6/ : FAILURE in 4m 43s\n- rally-tox-py36 http://logs.openstack.org/18/665918/1/check/rally-tox-py36/0d1f4cc/ : FAILURE in 3m 44s\n- rally-tox-py37 http://logs.openstack.org/18/665918/1/check/rally-tox-py37/55646de/ : FAILURE in 4m 18s\n- rally-tox-cover http://logs.openstack.org/18/665918/1/check/rally-tox-cover/858c17f/ : SUCCESS in 4m 57s\n- rally-tox-functional http://logs.openstack.org/18/665918/1/check/rally-tox-functional/00ced87/ : FAILURE in 6m 29s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/1/check/rally-tox-functional-py3/d47de1c/ : FAILURE in 6m 17s\n- rally-database-migration http://logs.openstack.org/18/665918/1/check/rally-database-migration/58c7320/ : SUCCESS in 8m 14s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/1/check/rally-install-ubuntu-xenial/2c782f5/ : SUCCESS in 4m 13s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/1/check/rally-install-ubuntu-bionic/8b4deed/ : SUCCESS in 3m 59s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/1/check/rally-install-centos-7/c932bbf/ : SUCCESS in 5m 35s\n- rally-tox-self http://logs.openstack.org/18/665918/1/check/rally-tox-self/0eb7bef/ : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9d32f54e116841926e131b1d74b2b40487cd9064","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-18 10:20:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b6e4d6fe42447fd52f5675b750c11f2db3c4716","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:57:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/2/check/openstack-tox-docs/1ee7e4d/html/ : SUCCESS in 3m 41s\n- rally-tox-docs http://logs.openstack.org/18/665918/2/check/rally-tox-docs/b946c85/ : SUCCESS in 3m 35s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/2/check/rally-tox-pep8/f01f0ce/ : SUCCESS in 4m 21s\n- rally-tox-py27 http://logs.openstack.org/18/665918/2/check/rally-tox-py27/b020822/ : SUCCESS in 3m 45s\n- rally-tox-py34 http://logs.openstack.org/18/665918/2/check/rally-tox-py34/4f455c6/ : SUCCESS in 4m 56s\n- rally-tox-py35 http://logs.openstack.org/18/665918/2/check/rally-tox-py35/16e4e12/ : SUCCESS in 4m 28s\n- rally-tox-py36 http://logs.openstack.org/18/665918/2/check/rally-tox-py36/c3d0780/ : SUCCESS in 4m 56s\n- rally-tox-py37 http://logs.openstack.org/18/665918/2/check/rally-tox-py37/f188713/ : SUCCESS in 4m 42s\n- rally-tox-cover http://logs.openstack.org/18/665918/2/check/rally-tox-cover/c04ae53/ : SUCCESS in 4m 19s\n- rally-tox-functional http://logs.openstack.org/18/665918/2/check/rally-tox-functional/4575453/ : SUCCESS in 6m 09s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/2/check/rally-tox-functional-py3/2690853/ : SUCCESS in 7m 30s\n- rally-database-migration http://logs.openstack.org/18/665918/2/check/rally-database-migration/02d313e/ : SUCCESS in 7m 55s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/2/check/rally-install-ubuntu-xenial/cf459bb/ : FAILURE in 3m 53s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/2/check/rally-install-ubuntu-bionic/05b7e62/ : SUCCESS in 4m 42s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/2/check/rally-install-centos-7/9532015/ : SUCCESS in 5m 46s\n- rally-tox-self http://logs.openstack.org/18/665918/2/check/rally-tox-self/cb8dd50/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"18e98d77b7731d4aae148f279a484fae8193f3ea","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-18 11:40:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a56694f48005569d005fbfb10a0dc65564835410","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-18 12:02:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2ea878718c3d10e6e9f6a09b639226ce89a93673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:26:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/4/check/openstack-tox-docs/c35bd6f/html/ : SUCCESS in 3m 32s\n- rally-tox-docs http://logs.openstack.org/18/665918/4/check/rally-tox-docs/5545a7c/ : SUCCESS in 4m 52s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/4/check/rally-tox-pep8/c4703b3/ : SUCCESS in 4m 05s\n- rally-tox-py27 http://logs.openstack.org/18/665918/4/check/rally-tox-py27/30b1f0d/ : SUCCESS in 3m 38s\n- rally-tox-py34 http://logs.openstack.org/18/665918/4/check/rally-tox-py34/230e6d9/ : SUCCESS in 5m 11s\n- rally-tox-py35 http://logs.openstack.org/18/665918/4/check/rally-tox-py35/937cd2f/ : SUCCESS in 4m 19s\n- rally-tox-py36 http://logs.openstack.org/18/665918/4/check/rally-tox-py36/e0fe43a/ : SUCCESS in 3m 33s\n- rally-tox-py37 http://logs.openstack.org/18/665918/4/check/rally-tox-py37/57e626d/ : SUCCESS in 4m 10s\n- rally-tox-cover http://logs.openstack.org/18/665918/4/check/rally-tox-cover/3330ce8/ : SUCCESS in 4m 17s\n- rally-tox-functional http://logs.openstack.org/18/665918/4/check/rally-tox-functional/b06bbb1/ : SUCCESS in 7m 13s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/4/check/rally-tox-functional-py3/3ee14e0/ : SUCCESS in 8m 20s\n- rally-database-migration http://logs.openstack.org/18/665918/4/check/rally-database-migration/0ad7962/ : SUCCESS in 7m 58s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/4/check/rally-install-ubuntu-xenial/22f05b6/ : FAILURE in 2m 56s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/4/check/rally-install-ubuntu-bionic/55c964f/ : SUCCESS in 3m 36s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/4/check/rally-install-centos-7/866addc/ : SUCCESS in 5m 08s\n- rally-tox-self http://logs.openstack.org/18/665918/4/check/rally-tox-self/531800a/ : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":4},{"id":"ef58f28a16381d592f606d82a65730d47d6d3ced","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-19 00:26:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"745e56ef7d75c00ecbbcd1f55bca288025dca51d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 00:48:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/5/check/openstack-tox-docs/831b265/html/ : SUCCESS in 3m 51s\n- rally-tox-docs http://logs.openstack.org/18/665918/5/check/rally-tox-docs/c78b8ca/ : SUCCESS in 4m 50s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/5/check/rally-tox-pep8/9abeb2a/ : SUCCESS in 4m 00s\n- rally-tox-py27 http://logs.openstack.org/18/665918/5/check/rally-tox-py27/4bdf187/ : SUCCESS in 3m 33s\n- rally-tox-py34 http://logs.openstack.org/18/665918/5/check/rally-tox-py34/d4c482d/ : SUCCESS in 4m 57s\n- rally-tox-py35 http://logs.openstack.org/18/665918/5/check/rally-tox-py35/0ab42f4/ : SUCCESS in 4m 32s\n- rally-tox-py36 http://logs.openstack.org/18/665918/5/check/rally-tox-py36/6109abf/ : SUCCESS in 3m 36s\n- rally-tox-py37 http://logs.openstack.org/18/665918/5/check/rally-tox-py37/4e1c367/ : SUCCESS in 3m 54s\n- rally-tox-cover http://logs.openstack.org/18/665918/5/check/rally-tox-cover/02be9b8/ : SUCCESS in 4m 29s\n- rally-tox-functional http://logs.openstack.org/18/665918/5/check/rally-tox-functional/4323e73/ : SUCCESS in 6m 12s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/5/check/rally-tox-functional-py3/00bc4e2/ : SUCCESS in 8m 07s\n- rally-database-migration http://logs.openstack.org/18/665918/5/check/rally-database-migration/fbc73b6/ : SUCCESS in 8m 05s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/5/check/rally-install-ubuntu-xenial/e6c55f8/ : SUCCESS in 3m 35s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/5/check/rally-install-ubuntu-bionic/d45855e/ : SUCCESS in 3m 21s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/5/check/rally-install-centos-7/b6ae240/ : SUCCESS in 5m 18s\n- rally-tox-self http://logs.openstack.org/18/665918/5/check/rally-tox-self/a9204f7/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":5},{"id":"1347d73405c68f90d581cef36e0fc3075f740b9e","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-06-19 10:28:57.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7f695663e2f2a01e8f4eb666634ab10aae457ce9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 11:14:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/6/check/openstack-tox-docs/7822150/html/ : SUCCESS in 3m 35s\n- rally-tox-docs http://logs.openstack.org/18/665918/6/check/rally-tox-docs/3f6b47e/ : SUCCESS in 3m 56s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/6/check/rally-tox-pep8/d587c70/ : SUCCESS in 4m 39s\n- rally-tox-py27 http://logs.openstack.org/18/665918/6/check/rally-tox-py27/4452af8/ : SUCCESS in 3m 53s\n- rally-tox-py34 http://logs.openstack.org/18/665918/6/check/rally-tox-py34/68ad085/ : SUCCESS in 5m 22s\n- rally-tox-py35 http://logs.openstack.org/18/665918/6/check/rally-tox-py35/f26a4d1/ : SUCCESS in 5m 06s\n- rally-tox-py36 http://logs.openstack.org/18/665918/6/check/rally-tox-py36/5432811/ : SUCCESS in 3m 35s\n- rally-tox-py37 http://logs.openstack.org/18/665918/6/check/rally-tox-py37/6337643/ : SUCCESS in 5m 40s\n- rally-tox-cover http://logs.openstack.org/18/665918/6/check/rally-tox-cover/98bb0b3/ : SUCCESS in 4m 24s\n- rally-tox-functional http://logs.openstack.org/18/665918/6/check/rally-tox-functional/639fb59/ : SUCCESS in 5m 58s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/6/check/rally-tox-functional-py3/2742549/ : SUCCESS in 5m 40s\n- rally-database-migration http://logs.openstack.org/18/665918/6/check/rally-database-migration/20ec4df/ : SUCCESS in 7m 59s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/6/check/rally-install-ubuntu-xenial/ae0c235/ : SUCCESS in 4m 45s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/6/check/rally-install-ubuntu-bionic/aea38db/ : SUCCESS in 3m 56s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/6/check/rally-install-centos-7/4f18438/ : SUCCESS in 5m 00s\n- rally-tox-self http://logs.openstack.org/18/665918/6/check/rally-tox-self/9f72bd8/ : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":6},{"id":"f8d6a4e11e4ccce8eca9b7041fd8603960d52ccf","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 13:19:50.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3945177ebe7e33ddb72125d48b8f21e67c69849c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 13:20:00.000000000","message":"Patch Set 6:\n\nthanks!","accounts_in_message":[],"_revision_number":6},{"id":"e176c26c9862eea5dad5297aba8594c434f16185","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:20:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4ce0a9f9c4d5c34924f547c7cf5ff687bd84c4bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:41:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/18/665918/6/gate/openstack-tox-docs/5e38d6f/html/ : SUCCESS in 3m 42s\n- rally-tox-docs http://logs.openstack.org/18/665918/6/gate/rally-tox-docs/bf8678b/ : SUCCESS in 4m 48s\n- rally-tox-pep8 http://logs.openstack.org/18/665918/6/gate/rally-tox-pep8/cbe887d/ : SUCCESS in 3m 55s\n- rally-tox-py27 http://logs.openstack.org/18/665918/6/gate/rally-tox-py27/b41a0b4/ : SUCCESS in 3m 57s\n- rally-tox-py34 http://logs.openstack.org/18/665918/6/gate/rally-tox-py34/bcbb25b/ : SUCCESS in 4m 57s\n- rally-tox-py35 http://logs.openstack.org/18/665918/6/gate/rally-tox-py35/3c4b216/ : SUCCESS in 4m 49s\n- rally-tox-py36 http://logs.openstack.org/18/665918/6/gate/rally-tox-py36/872ca96/ : SUCCESS in 3m 37s\n- rally-tox-py37 http://logs.openstack.org/18/665918/6/gate/rally-tox-py37/e19b17e/ : SUCCESS in 4m 55s\n- rally-tox-cover http://logs.openstack.org/18/665918/6/gate/rally-tox-cover/10357bb/ : SUCCESS in 4m 07s\n- rally-tox-functional http://logs.openstack.org/18/665918/6/gate/rally-tox-functional/4eebc90/ : SUCCESS in 5m 48s\n- rally-tox-functional-py3 http://logs.openstack.org/18/665918/6/gate/rally-tox-functional-py3/c7a87b1/ : SUCCESS in 5m 51s\n- rally-database-migration http://logs.openstack.org/18/665918/6/gate/rally-database-migration/26654ad/ : SUCCESS in 7m 43s\n- rally-install-ubuntu-xenial http://logs.openstack.org/18/665918/6/gate/rally-install-ubuntu-xenial/54735aa/ : SUCCESS in 3m 42s\n- rally-install-ubuntu-bionic http://logs.openstack.org/18/665918/6/gate/rally-install-ubuntu-bionic/8e90bac/ : SUCCESS in 4m 23s\n- rally-install-centos-7 http://logs.openstack.org/18/665918/6/gate/rally-install-centos-7/4293688/ : SUCCESS in 5m 09s\n- rally-tox-self http://logs.openstack.org/18/665918/6/gate/rally-tox-self/055896a/ : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":6},{"id":"159f3e774d107da62f2c0819fdafd8350cbf3bf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:41:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c5a5bc83de0bde967b00b012f5b6e76d3416a1b9","revisions":{"adfc277c136ae27fef7fe0795cc9ea5e9434e792":{"kind":"REWORK","_number":1,"created":"2019-06-18 09:04:22.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/1"}}},"commit":{"parents":[{"commit":"acc2aa276fd6542ebac2b14bdf43bdeec12c6dab","subject":"Add some tests for verify function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/acc2aa276fd6542ebac2b14bdf43bdeec12c6dab"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"subject":"TEST `rally verify tempest` for rally-openstack","message":"TEST `rally verify tempest` for rally-openstack\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/adfc277c136ae27fef7fe0795cc9ea5e9434e792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/adfc277c136ae27fef7fe0795cc9ea5e9434e792"}]},"branch":"refs/heads/master"},"84b4a9fc678c0192569988526edf3a48590867f6":{"kind":"REWORK","_number":2,"created":"2019-06-18 10:20:13.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/2"}}},"commit":{"parents":[{"commit":"acc2aa276fd6542ebac2b14bdf43bdeec12c6dab","subject":"Add some tests for verify function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/acc2aa276fd6542ebac2b14bdf43bdeec12c6dab"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 10:43:21.000000000","tz":480},"subject":"TEST `rally verify tempest` for rally-openstack","message":"TEST `rally verify tempest` for rally-openstack\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/84b4a9fc678c0192569988526edf3a48590867f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/84b4a9fc678c0192569988526edf3a48590867f6"}]},"branch":"refs/heads/master"},"a274786e9431298463eef3694d54d6992e0f25a7":{"kind":"REWORK","_number":3,"created":"2019-06-18 11:40:07.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/3"}}},"commit":{"parents":[{"commit":"acc2aa276fd6542ebac2b14bdf43bdeec12c6dab","subject":"Add some tests for verify function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/acc2aa276fd6542ebac2b14bdf43bdeec12c6dab"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 12:05:27.000000000","tz":480},"subject":"Update upper-constraints.txt","message":"Update upper-constraints.txt\n\nupdate upper-constraints.txt  and Fix some issues since\nSQLAlchemy was upgraded to 1.3.4\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a274786e9431298463eef3694d54d6992e0f25a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a274786e9431298463eef3694d54d6992e0f25a7"}]},"branch":"refs/heads/master"},"3a8b4507baf84b6dc07e187e60b45cce14c950c1":{"kind":"REWORK","_number":4,"created":"2019-06-18 12:02:35.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/4"}}},"commit":{"parents":[{"commit":"acc2aa276fd6542ebac2b14bdf43bdeec12c6dab","subject":"Add some tests for verify function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/acc2aa276fd6542ebac2b14bdf43bdeec12c6dab"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 12:28:53.000000000","tz":480},"subject":"Update SQLAlchemy in  upper-constraints.txt","message":"Update SQLAlchemy in  upper-constraints.txt\n\nupdate SQLAlchemy to be compatible with openstack\nand Fix some issues since SQLAlchemy was upgraded to 1.3.4\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3a8b4507baf84b6dc07e187e60b45cce14c950c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3a8b4507baf84b6dc07e187e60b45cce14c950c1"}]},"branch":"refs/heads/master"},"6d924e1a6cf621d9ffefd6c4258f334c51ff261f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-19 00:26:08.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/5"}}},"commit":{"parents":[{"commit":"9c64f964e6e50c2a3ea63d2725e5c4839756c32d","subject":"Add some tests for verify function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/9c64f964e6e50c2a3ea63d2725e5c4839756c32d"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-19 00:26:08.000000000","tz":0},"subject":"Update SQLAlchemy in  upper-constraints.txt","message":"Update SQLAlchemy in  upper-constraints.txt\n\nupdate SQLAlchemy to be compatible with openstack\nand Fix some issues since SQLAlchemy was upgraded to 1.3.4\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6d924e1a6cf621d9ffefd6c4258f334c51ff261f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6d924e1a6cf621d9ffefd6c4258f334c51ff261f"}]},"branch":"refs/heads/master"},"c5a5bc83de0bde967b00b012f5b6e76d3416a1b9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-19 10:28:57.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/18/665918/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/18/665918/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/18/665918/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/18/665918/6"}}},"commit":{"parents":[{"commit":"bcdcfc7900aad26635df793c90b36d3f2e6bf386","subject":"install_rally.sh: moved zypper pkg manager up to handle case when yum is installed on SUSE to fulfill dependencies.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bcdcfc7900aad26635df793c90b36d3f2e6bf386"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-18 09:30:29.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-06-19 10:56:44.000000000","tz":480},"subject":"Update SQLAlchemy in  upper-constraints.txt","message":"Update SQLAlchemy in  upper-constraints.txt\n\nupdate SQLAlchemy to be compatible with openstack\nand Fix some issues since SQLAlchemy was upgraded to 1.3.4\n\nChange-Id: Id1955160813c9be87dffc7e38b8d3476cec04d60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c5a5bc83de0bde967b00b012f5b6e76d3416a1b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c5a5bc83de0bde967b00b012f5b6e76d3416a1b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
