)]}'
{"id":"openstack%2Frally~666312","triplet_id":"openstack%2Frally~master~I8ac7f5560b0c0d51bf744e1be6bf3768c978e5df","project":"openstack/rally","branch":"master","topic":"verifier","hashtags":[],"change_id":"I8ac7f5560b0c0d51bf744e1be6bf3768c978e5df","subject":"[verify ]Fix several py3 issues","status":"MERGED","created":"2019-06-19 15:04:13.000000000","updated":"2019-06-19 15:50:49.000000000","submitted":"2019-06-19 15:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666312-1560959449284-96f6c04a","meta_rev_id":"3b740862efe81f54f922b78247e2994abb518d80","_number":666312,"virtual_id_number":666312,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-19 15:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":2,"date":"2019-06-19 15:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":0,"date":"2019-06-19 15:04:13.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"value":1,"date":"2019-06-19 15:04:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 15:04:13.000000000","updated_by":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"reviewer":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"state":"REVIEWER"},{"updated":"2019-06-19 15:50:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"84a98164d5e409f6bc5848691461f007c71066a3","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 15:04:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7308902f5ec42dfe2205364dacf901f4fae207d3","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 15:04:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3acc5f1175d25a5704a3860032a6e9007f3aff54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:34:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/666312/1/check/openstack-tox-docs/d8b16a1/html/ : SUCCESS in 3m 24s\n- rally-tox-docs http://logs.openstack.org/12/666312/1/check/rally-tox-docs/c1ba705/ : SUCCESS in 3m 44s\n- rally-tox-pep8 http://logs.openstack.org/12/666312/1/check/rally-tox-pep8/ed3eb0a/ : SUCCESS in 4m 33s\n- rally-tox-py27 http://logs.openstack.org/12/666312/1/check/rally-tox-py27/89405c2/ : SUCCESS in 4m 10s\n- rally-tox-py34 http://logs.openstack.org/12/666312/1/check/rally-tox-py34/9639980/ : SUCCESS in 4m 44s\n- rally-tox-py35 http://logs.openstack.org/12/666312/1/check/rally-tox-py35/d6eea59/ : SUCCESS in 4m 28s\n- rally-tox-py36 http://logs.openstack.org/12/666312/1/check/rally-tox-py36/0095c2f/ : SUCCESS in 4m 09s\n- rally-tox-py37 http://logs.openstack.org/12/666312/1/check/rally-tox-py37/6bd9915/ : SUCCESS in 4m 12s\n- rally-tox-cover http://logs.openstack.org/12/666312/1/check/rally-tox-cover/3cc3c65/ : SUCCESS in 4m 42s\n- rally-tox-functional http://logs.openstack.org/12/666312/1/check/rally-tox-functional/83286bc/ : SUCCESS in 8m 20s\n- rally-tox-functional-py3 http://logs.openstack.org/12/666312/1/check/rally-tox-functional-py3/066ee6a/ : SUCCESS in 6m 10s\n- rally-database-migration http://logs.openstack.org/12/666312/1/check/rally-database-migration/bd1e278/ : SUCCESS in 7m 35s\n- rally-install-ubuntu-xenial http://logs.openstack.org/12/666312/1/check/rally-install-ubuntu-xenial/1aff55c/ : SUCCESS in 4m 14s\n- rally-install-ubuntu-bionic http://logs.openstack.org/12/666312/1/check/rally-install-ubuntu-bionic/89e5a16/ : SUCCESS in 3m 40s\n- rally-install-centos-7 http://logs.openstack.org/12/666312/1/check/rally-install-centos-7/6b47a32/ : SUCCESS in 5m 31s\n- rally-tox-self http://logs.openstack.org/12/666312/1/check/rally-tox-self/4d75108/ : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f5fdcbc9ac14e2c0d727e27fe0a29c262c15025d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:34:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2b73eccd33f5d61735491720ef0b3945ccee5bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:50:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/12/666312/1/gate/openstack-tox-docs/a9f1cb9/html/ : SUCCESS in 3m 20s\n- rally-tox-docs http://logs.openstack.org/12/666312/1/gate/rally-tox-docs/9116870/ : SUCCESS in 3m 53s\n- rally-tox-pep8 http://logs.openstack.org/12/666312/1/gate/rally-tox-pep8/db28e11/ : SUCCESS in 3m 55s\n- rally-tox-py27 http://logs.openstack.org/12/666312/1/gate/rally-tox-py27/ba1e9f7/ : SUCCESS in 4m 06s\n- rally-tox-py34 http://logs.openstack.org/12/666312/1/gate/rally-tox-py34/9470b04/ : SUCCESS in 4m 45s\n- rally-tox-py35 http://logs.openstack.org/12/666312/1/gate/rally-tox-py35/fd42b38/ : SUCCESS in 4m 34s\n- rally-tox-py36 http://logs.openstack.org/12/666312/1/gate/rally-tox-py36/83cd4d8/ : SUCCESS in 3m 59s\n- rally-tox-py37 http://logs.openstack.org/12/666312/1/gate/rally-tox-py37/da5d37a/ : SUCCESS in 4m 43s\n- rally-tox-cover http://logs.openstack.org/12/666312/1/gate/rally-tox-cover/273d9fa/ : SUCCESS in 4m 27s\n- rally-tox-functional http://logs.openstack.org/12/666312/1/gate/rally-tox-functional/86720ee/ : SUCCESS in 5m 37s\n- rally-tox-functional-py3 http://logs.openstack.org/12/666312/1/gate/rally-tox-functional-py3/5f2c1f0/ : SUCCESS in 5m 58s\n- rally-database-migration http://logs.openstack.org/12/666312/1/gate/rally-database-migration/06722c6/ : SUCCESS in 8m 27s\n- rally-install-ubuntu-xenial http://logs.openstack.org/12/666312/1/gate/rally-install-ubuntu-xenial/ae01bd4/ : SUCCESS in 3m 58s\n- rally-install-ubuntu-bionic http://logs.openstack.org/12/666312/1/gate/rally-install-ubuntu-bionic/3d18246/ : SUCCESS in 4m 08s\n- rally-install-centos-7 http://logs.openstack.org/12/666312/1/gate/rally-install-centos-7/00625ef/ : SUCCESS in 5m 10s\n- rally-tox-self http://logs.openstack.org/12/666312/1/gate/rally-tox-self/ef91913/ : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4d676d5a5ef2b0f95bd88574498e01f5cc616e1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:50:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45cc9ad01f0eaa06735e5c8518bac80143a854f9","revisions":{"45cc9ad01f0eaa06735e5c8518bac80143a854f9":{"kind":"REWORK","_number":1,"created":"2019-06-19 15:04:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/12/666312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/12/666312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/666312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/666312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/12/666312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/12/666312/1"}}},"commit":{"parents":[{"commit":"c5a5bc83de0bde967b00b012f5b6e76d3416a1b9","subject":"Update SQLAlchemy in  upper-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c5a5bc83de0bde967b00b012f5b6e76d3416a1b9"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-04-19 03:56:43.000000000","tz":480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-06-19 15:03:26.000000000","tz":180},"subject":"[verify ]Fix several py3 issues","message":"[verify ]Fix several py3 issues\n\n1. switch \\u2714 to :-)\n   a. \\u2714 can not be decoded on py3\n   b. \\u2714 can not shown rightly on Window as Linux\n\n2. compatiable py3 and py2 for load_lists\n\nChange-Id: I8ac7f5560b0c0d51bf744e1be6bf3768c978e5df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45cc9ad01f0eaa06735e5c8518bac80143a854f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45cc9ad01f0eaa06735e5c8518bac80143a854f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
