)]}'
{"id":"openstack%2Frally~666314","triplet_id":"openstack%2Frally~master~I803cde8b0c498f81f26af7c8f710ef9b6ae8f11f","project":"openstack/rally","branch":"master","topic":"newRelease","hashtags":[],"change_id":"I803cde8b0c498f81f26af7c8f710ef9b6ae8f11f","subject":"Propose 1.6.0 release","status":"MERGED","created":"2019-06-19 15:11:38.000000000","updated":"2019-06-19 15:58:06.000000000","submitted":"2019-06-19 15:58:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666314-1560959886161-6407a728","meta_rev_id":"93822d5db26d3e3c05c3ce14a6b2555e2d1fdf10","_number":666314,"virtual_id_number":666314,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-19 15:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-06-19 15:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-06-19 15:13:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 15:58:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab6945f23a7c235daa59634cba382e90a6e94389","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 15:11:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd192d623b97bd3a7a7bfa0ecebc6518054dd970","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 15:13:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"93d2409b14d2e3b4e3bec7e5b39e27174c6c45af","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-19 15:13:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"23a3fcf6b890ecdd49162a50e05b64606b2a7901","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:27:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/666314/2/check/openstack-tox-docs/0f2ad78/html/ : SUCCESS in 3m 30s\n- rally-tox-docs http://logs.openstack.org/14/666314/2/check/rally-tox-docs/841dff7/ : SUCCESS in 3m 35s\n- rally-tox-pep8 http://logs.openstack.org/14/666314/2/check/rally-tox-pep8/37720d1/ : SUCCESS in 3m 39s\n- rally-tox-py27 http://logs.openstack.org/14/666314/2/check/rally-tox-py27/b0c34ba/ : SUCCESS in 3m 43s\n- rally-tox-py34 http://logs.openstack.org/14/666314/2/check/rally-tox-py34/5437ab7/ : SUCCESS in 4m 39s\n- rally-tox-py35 http://logs.openstack.org/14/666314/2/check/rally-tox-py35/5b26848/ : SUCCESS in 4m 28s\n- rally-tox-py36 http://logs.openstack.org/14/666314/2/check/rally-tox-py36/db35832/ : SUCCESS in 4m 12s\n- rally-tox-py37 http://logs.openstack.org/14/666314/2/check/rally-tox-py37/91e8cac/ : SUCCESS in 4m 14s\n- rally-tox-cover http://logs.openstack.org/14/666314/2/check/rally-tox-cover/30e434e/ : SUCCESS in 4m 28s\n- rally-tox-functional http://logs.openstack.org/14/666314/2/check/rally-tox-functional/141e626/ : SUCCESS in 5m 53s\n- rally-tox-functional-py3 http://logs.openstack.org/14/666314/2/check/rally-tox-functional-py3/85e9032/ : SUCCESS in 6m 13s\n- rally-database-migration http://logs.openstack.org/14/666314/2/check/rally-database-migration/d5be336/ : SUCCESS in 7m 58s\n- rally-install-ubuntu-xenial http://logs.openstack.org/14/666314/2/check/rally-install-ubuntu-xenial/cc91ff5/ : SUCCESS in 4m 11s\n- rally-install-ubuntu-bionic http://logs.openstack.org/14/666314/2/check/rally-install-ubuntu-bionic/76ef3fc/ : SUCCESS in 3m 24s\n- rally-install-centos-7 http://logs.openstack.org/14/666314/2/check/rally-install-centos-7/3e8b256/ : SUCCESS in 4m 19s\n- rally-tox-self http://logs.openstack.org/14/666314/2/check/rally-tox-self/24afad1/ : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e0e6a8c1016fdcb5a3a3852da19b78ffe7672d1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:34:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d2b20d7f9d6cc3f2690c6e5adcb644c647a05c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:58:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/14/666314/2/gate/openstack-tox-docs/9f8e1b0/html/ : SUCCESS in 3m 36s\n- rally-tox-docs http://logs.openstack.org/14/666314/2/gate/rally-tox-docs/d4765c5/ : SUCCESS in 4m 04s\n- rally-tox-pep8 http://logs.openstack.org/14/666314/2/gate/rally-tox-pep8/d1fcc1a/ : SUCCESS in 4m 22s\n- rally-tox-py27 http://logs.openstack.org/14/666314/2/gate/rally-tox-py27/2db1980/ : SUCCESS in 4m 00s\n- rally-tox-py34 http://logs.openstack.org/14/666314/2/gate/rally-tox-py34/f61450f/ : SUCCESS in 4m 31s\n- rally-tox-py35 http://logs.openstack.org/14/666314/2/gate/rally-tox-py35/6637a4b/ : SUCCESS in 5m 11s\n- rally-tox-py36 http://logs.openstack.org/14/666314/2/gate/rally-tox-py36/3c8f192/ : SUCCESS in 4m 02s\n- rally-tox-py37 http://logs.openstack.org/14/666314/2/gate/rally-tox-py37/4a2d4da/ : SUCCESS in 4m 25s\n- rally-tox-cover http://logs.openstack.org/14/666314/2/gate/rally-tox-cover/08dc3dd/ : SUCCESS in 4m 13s\n- rally-tox-functional http://logs.openstack.org/14/666314/2/gate/rally-tox-functional/609c37c/ : SUCCESS in 5m 43s\n- rally-tox-functional-py3 http://logs.openstack.org/14/666314/2/gate/rally-tox-functional-py3/182168f/ : SUCCESS in 6m 34s\n- rally-database-migration http://logs.openstack.org/14/666314/2/gate/rally-database-migration/13fb0dc/ : SUCCESS in 7m 42s\n- rally-install-ubuntu-xenial http://logs.openstack.org/14/666314/2/gate/rally-install-ubuntu-xenial/51c25f5/ : SUCCESS in 3m 44s\n- rally-install-ubuntu-bionic http://logs.openstack.org/14/666314/2/gate/rally-install-ubuntu-bionic/2919df6/ : SUCCESS in 3m 50s\n- rally-install-centos-7 http://logs.openstack.org/14/666314/2/gate/rally-install-centos-7/45b4914/ : SUCCESS in 5m 55s\n- rally-tox-self http://logs.openstack.org/14/666314/2/gate/rally-tox-self/f5a7b93/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b3ae7a9109a78f78713f949ce7cf1fd6ae32a62c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 15:58:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee13ba4983c86bb21096a14a681593f958a416a2","revisions":{"1eeb4ce05bf9860aca9a0fd605e24500dff6ae04":{"kind":"REWORK","_number":1,"created":"2019-06-19 15:11:38.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/14/666314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/14/666314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/14/666314/1"}}},"commit":{"parents":[{"commit":"c5a5bc83de0bde967b00b012f5b6e76d3416a1b9","subject":"Update SQLAlchemy in  upper-constraints.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c5a5bc83de0bde967b00b012f5b6e76d3416a1b9"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-06-19 15:11:30.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-06-19 15:11:30.000000000","tz":180},"subject":"Propose 1.6.0 release","message":"Propose 1.6.0 release\n\nChange-Id: I803cde8b0c498f81f26af7c8f710ef9b6ae8f11f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1eeb4ce05bf9860aca9a0fd605e24500dff6ae04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1eeb4ce05bf9860aca9a0fd605e24500dff6ae04"}]},"branch":"refs/heads/master"},"ee13ba4983c86bb21096a14a681593f958a416a2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-19 15:13:19.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/14/666314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/14/666314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/14/666314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/14/666314/2"}}},"commit":{"parents":[{"commit":"45cc9ad01f0eaa06735e5c8518bac80143a854f9","subject":"[verify ]Fix several py3 issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/45cc9ad01f0eaa06735e5c8518bac80143a854f9"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-06-19 15:11:30.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-06-19 15:13:19.000000000","tz":0},"subject":"Propose 1.6.0 release","message":"Propose 1.6.0 release\n\nChange-Id: I803cde8b0c498f81f26af7c8f710ef9b6ae8f11f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee13ba4983c86bb21096a14a681593f958a416a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee13ba4983c86bb21096a14a681593f958a416a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
