)]}'
{"id":"openstack%2Frally~667008","triplet_id":"openstack%2Frally~master~I99faabd9a299a5cae6154b371740cefc055f333e","project":"openstack/rally","branch":"master","topic":"asset","hashtags":[],"change_id":"I99faabd9a299a5cae6154b371740cefc055f333e","subject":"Fix wrong assert function name","status":"MERGED","created":"2019-06-24 04:17:46.000000000","updated":"2019-06-24 08:51:21.000000000","submitted":"2019-06-24 08:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667008-1561366281242-bfd37fb7","meta_rev_id":"08b5e987a883f57c321c13522e35e7cfbb5a8c36","_number":667008,"virtual_id_number":667008,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-24 08:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-06-24 08:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-06-24 08:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-24 08:38:49.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-06-24 08:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f20388ddb822a1e7b6e65fe672278fbaf27d4bc","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-24 04:17:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cec7696c05b50a5ad9db94034f3fe02ecae7e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 04:30:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/667008/1/check/openstack-tox-docs/08f29cf/html/ : SUCCESS in 3m 42s\n- rally-tox-docs http://logs.openstack.org/08/667008/1/check/rally-tox-docs/c523421/ : SUCCESS in 3m 38s\n- rally-tox-pep8 http://logs.openstack.org/08/667008/1/check/rally-tox-pep8/fc1d944/ : SUCCESS in 3m 51s\n- rally-tox-py27 http://logs.openstack.org/08/667008/1/check/rally-tox-py27/08c02a1/ : SUCCESS in 4m 02s\n- rally-tox-py34 http://logs.openstack.org/08/667008/1/check/rally-tox-py34/444168b/ : SUCCESS in 4m 23s\n- rally-tox-py35 http://logs.openstack.org/08/667008/1/check/rally-tox-py35/6c368a7/ : SUCCESS in 4m 04s\n- rally-tox-py36 http://logs.openstack.org/08/667008/1/check/rally-tox-py36/a20b80a/ : SUCCESS in 3m 28s\n- rally-tox-py37 http://logs.openstack.org/08/667008/1/check/rally-tox-py37/acaa6c2/ : SUCCESS in 4m 07s\n- rally-tox-cover http://logs.openstack.org/08/667008/1/check/rally-tox-cover/f5bc58d/ : SUCCESS in 4m 01s\n- rally-tox-functional http://logs.openstack.org/08/667008/1/check/rally-tox-functional/2203da6/ : SUCCESS in 5m 38s\n- rally-tox-functional-py3 http://logs.openstack.org/08/667008/1/check/rally-tox-functional-py3/c60a52b/ : SUCCESS in 5m 42s\n- rally-database-migration http://logs.openstack.org/08/667008/1/check/rally-database-migration/335b582/ : SUCCESS in 7m 31s\n- rally-install-ubuntu-xenial http://logs.openstack.org/08/667008/1/check/rally-install-ubuntu-xenial/1d1115d/ : SUCCESS in 3m 48s\n- rally-install-ubuntu-bionic http://logs.openstack.org/08/667008/1/check/rally-install-ubuntu-bionic/b53d566/ : SUCCESS in 3m 10s\n- rally-install-centos-7 http://logs.openstack.org/08/667008/1/check/rally-install-centos-7/ba6f2c1/ : SUCCESS in 5m 16s\n- rally-tox-self http://logs.openstack.org/08/667008/1/check/rally-tox-self/f03b14e/ : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"11d5b541e2da40e16253df3c74853b4e32933cd7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-24 08:38:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f33b2cf3050915370a5136430f8cc8549a2e763","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:39:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70db96fbaa46d699d82298d9a8f747b2bba02d88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"15b84afca1f975d75c5bad951bd2d0217f431446","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:51:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/667008/1/gate/openstack-tox-docs/ecec77b/html/ : SUCCESS in 3m 24s\n- rally-tox-docs http://logs.openstack.org/08/667008/1/gate/rally-tox-docs/fe9cab5/ : SUCCESS in 4m 05s\n- rally-tox-pep8 http://logs.openstack.org/08/667008/1/gate/rally-tox-pep8/1cc5153/ : SUCCESS in 4m 15s\n- rally-tox-py27 http://logs.openstack.org/08/667008/1/gate/rally-tox-py27/1739f79/ : SUCCESS in 4m 02s\n- rally-tox-py34 http://logs.openstack.org/08/667008/1/gate/rally-tox-py34/1860393/ : SUCCESS in 4m 37s\n- rally-tox-py35 http://logs.openstack.org/08/667008/1/gate/rally-tox-py35/b048930/ : SUCCESS in 4m 31s\n- rally-tox-py36 http://logs.openstack.org/08/667008/1/gate/rally-tox-py36/1f8d90b/ : SUCCESS in 3m 36s\n- rally-tox-py37 http://logs.openstack.org/08/667008/1/gate/rally-tox-py37/948a63a/ : SUCCESS in 4m 32s\n- rally-tox-cover http://logs.openstack.org/08/667008/1/gate/rally-tox-cover/cdf2886/ : SUCCESS in 4m 48s\n- rally-tox-functional http://logs.openstack.org/08/667008/1/gate/rally-tox-functional/53ba1c2/ : SUCCESS in 6m 04s\n- rally-tox-functional-py3 http://logs.openstack.org/08/667008/1/gate/rally-tox-functional-py3/cd2ad78/ : SUCCESS in 6m 17s\n- rally-database-migration http://logs.openstack.org/08/667008/1/gate/rally-database-migration/28bb774/ : SUCCESS in 7m 32s\n- rally-install-ubuntu-xenial http://logs.openstack.org/08/667008/1/gate/rally-install-ubuntu-xenial/e6211de/ : SUCCESS in 4m 54s\n- rally-install-ubuntu-bionic http://logs.openstack.org/08/667008/1/gate/rally-install-ubuntu-bionic/579d277/ : SUCCESS in 3m 48s\n- rally-install-centos-7 http://logs.openstack.org/08/667008/1/gate/rally-install-centos-7/c0058ca/ : SUCCESS in 5m 15s\n- rally-tox-self http://logs.openstack.org/08/667008/1/gate/rally-tox-self/db1d18d/ : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0cfbe2def0ec8e9cf77136abf8189b56b2b585b","revisions":{"a0cfbe2def0ec8e9cf77136abf8189b56b2b585b":{"kind":"REWORK","_number":1,"created":"2019-06-24 04:17:46.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/08/667008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/08/667008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/08/667008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/08/667008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/08/667008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/08/667008/1"}}},"commit":{"parents":[{"commit":"ee13ba4983c86bb21096a14a681593f958a416a2","subject":"Propose 1.6.0 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ee13ba4983c86bb21096a14a681593f958a416a2"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-24 04:11:49.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-24 04:11:49.000000000","tz":480},"subject":"Fix wrong assert function name","message":"Fix wrong assert function name\n\nThis is to fix wrong assert function name of asset_called_once_with.\n\nChange-Id: I99faabd9a299a5cae6154b371740cefc055f333e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a0cfbe2def0ec8e9cf77136abf8189b56b2b585b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a0cfbe2def0ec8e9cf77136abf8189b56b2b585b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
