)]}'
{"id":"openstack%2Frally~667777","triplet_id":"openstack%2Frally~master~I8ce80b7b93d4846219ad886b8531daef56e16fa3","project":"openstack/rally","branch":"master","topic":"asser","hashtags":[],"change_id":"I8ce80b7b93d4846219ad886b8531daef56e16fa3","subject":"Fix wrong assert function name","status":"MERGED","created":"2019-06-27 02:56:35.000000000","updated":"2019-06-27 14:40:23.000000000","submitted":"2019-06-27 14:40:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667777-1561646423450-d253eb7f","meta_rev_id":"5f22519a0dc863a406c560dd7f300cce0747ab79","_number":667777,"virtual_id_number":667777,"owner":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-06-27 14:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-06-27 14:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-06-27 14:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 14:18:55.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-06-27 14:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1bca3c5f273e428ece35a755c76fbdb0fb396a7","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-27 02:56:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b60a657feb4004bf5c34d535a18efe61483dcc3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:10:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/77/667777/1/check/openstack-tox-docs/97e3288/html/ : SUCCESS in 3m 14s\n- rally-tox-docs http://logs.openstack.org/77/667777/1/check/rally-tox-docs/d32992a/ : SUCCESS in 4m 10s\n- rally-tox-pep8 http://logs.openstack.org/77/667777/1/check/rally-tox-pep8/cf4a18e/ : SUCCESS in 4m 30s\n- rally-tox-py27 http://logs.openstack.org/77/667777/1/check/rally-tox-py27/f839169/ : FAILURE in 4m 23s\n- rally-tox-py34 http://logs.openstack.org/77/667777/1/check/rally-tox-py34/40708de/ : FAILURE in 4m 55s\n- rally-tox-py35 http://logs.openstack.org/77/667777/1/check/rally-tox-py35/07a961e/ : FAILURE in 4m 15s\n- rally-tox-py36 http://logs.openstack.org/77/667777/1/check/rally-tox-py36/adb0a9f/ : FAILURE in 3m 56s\n- rally-tox-py37 http://logs.openstack.org/77/667777/1/check/rally-tox-py37/079849c/ : FAILURE in 4m 18s\n- rally-tox-cover http://logs.openstack.org/77/667777/1/check/rally-tox-cover/9a78141/ : SUCCESS in 3m 55s\n- rally-tox-functional http://logs.openstack.org/77/667777/1/check/rally-tox-functional/8890391/ : SUCCESS in 5m 37s\n- rally-tox-functional-py3 http://logs.openstack.org/77/667777/1/check/rally-tox-functional-py3/0e7375d/ : SUCCESS in 5m 56s\n- rally-database-migration http://logs.openstack.org/77/667777/1/check/rally-database-migration/f2d3cf9/ : SUCCESS in 7m 19s\n- rally-install-ubuntu-xenial http://logs.openstack.org/77/667777/1/check/rally-install-ubuntu-xenial/b4f3d5b/ : SUCCESS in 3m 43s\n- rally-install-ubuntu-bionic http://logs.openstack.org/77/667777/1/check/rally-install-ubuntu-bionic/9bddcd8/ : SUCCESS in 3m 42s\n- rally-install-centos-7 http://logs.openstack.org/77/667777/1/check/rally-install-centos-7/b66b878/ : SUCCESS in 5m 18s\n- rally-tox-self http://logs.openstack.org/77/667777/1/check/rally-tox-self/cdcd6fb/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dc6aa0d60c901b9c248f60ca73f14a0ddd41cd58","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-27 03:19:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"53818eeb011ef70dcc1cafe296e2067d6f847d11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:38:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/77/667777/2/check/openstack-tox-docs/bbb9692/html/ : SUCCESS in 3m 39s\n- rally-tox-docs http://logs.openstack.org/77/667777/2/check/rally-tox-docs/6131532/ : SUCCESS in 4m 10s\n- rally-tox-pep8 http://logs.openstack.org/77/667777/2/check/rally-tox-pep8/08b1da6/ : SUCCESS in 4m 21s\n- rally-tox-py27 http://logs.openstack.org/77/667777/2/check/rally-tox-py27/9bf0c51/ : FAILURE in 4m 46s\n- rally-tox-py34 http://logs.openstack.org/77/667777/2/check/rally-tox-py34/9aa417e/ : FAILURE in 4m 49s\n- rally-tox-py35 http://logs.openstack.org/77/667777/2/check/rally-tox-py35/3ce622b/ : FAILURE in 3m 54s\n- rally-tox-py36 http://logs.openstack.org/77/667777/2/check/rally-tox-py36/98cb265/ : FAILURE in 3m 41s\n- rally-tox-py37 http://logs.openstack.org/77/667777/2/check/rally-tox-py37/408f622/ : FAILURE in 5m 35s\n- rally-tox-cover http://logs.openstack.org/77/667777/2/check/rally-tox-cover/cb86482/ : SUCCESS in 4m 08s\n- rally-tox-functional http://logs.openstack.org/77/667777/2/check/rally-tox-functional/6ac456b/ : SUCCESS in 5m 50s\n- rally-tox-functional-py3 http://logs.openstack.org/77/667777/2/check/rally-tox-functional-py3/1eaeea0/ : SUCCESS in 5m 35s\n- rally-database-migration http://logs.openstack.org/77/667777/2/check/rally-database-migration/036049a/ : SUCCESS in 7m 57s\n- rally-install-ubuntu-xenial http://logs.openstack.org/77/667777/2/check/rally-install-ubuntu-xenial/f4a48db/ : SUCCESS in 3m 32s\n- rally-install-ubuntu-bionic http://logs.openstack.org/77/667777/2/check/rally-install-ubuntu-bionic/b898d90/ : SUCCESS in 4m 40s\n- rally-install-centos-7 http://logs.openstack.org/77/667777/2/check/rally-install-centos-7/97e2278/ : SUCCESS in 5m 13s\n- rally-tox-self http://logs.openstack.org/77/667777/2/check/rally-tox-self/3ed3b6f/ : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":2},{"id":"df4af1fb1a3ca33e3f1522edf07a45de3c3d086e","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2019-06-27 03:54:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e2a1da5076049d9214c72dafe768e1dd993776b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:16:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/667777/3/check/openstack-tox-docs/46dc577/html/ : SUCCESS in 2m 58s\n- rally-tox-docs http://logs.openstack.org/77/667777/3/check/rally-tox-docs/a2288ce/ : SUCCESS in 3m 31s\n- rally-tox-pep8 http://logs.openstack.org/77/667777/3/check/rally-tox-pep8/65efffb/ : SUCCESS in 3m 52s\n- rally-tox-py27 http://logs.openstack.org/77/667777/3/check/rally-tox-py27/1082cf9/ : SUCCESS in 4m 04s\n- rally-tox-py34 http://logs.openstack.org/77/667777/3/check/rally-tox-py34/326fffc/ : SUCCESS in 5m 05s\n- rally-tox-py35 http://logs.openstack.org/77/667777/3/check/rally-tox-py35/470a34a/ : SUCCESS in 5m 15s\n- rally-tox-py36 http://logs.openstack.org/77/667777/3/check/rally-tox-py36/d83770b/ : SUCCESS in 5m 06s\n- rally-tox-py37 http://logs.openstack.org/77/667777/3/check/rally-tox-py37/7dd9151/ : SUCCESS in 3m 52s\n- rally-tox-cover http://logs.openstack.org/77/667777/3/check/rally-tox-cover/517c8ba/ : SUCCESS in 4m 03s\n- rally-tox-functional http://logs.openstack.org/77/667777/3/check/rally-tox-functional/92d71bb/ : SUCCESS in 6m 00s\n- rally-tox-functional-py3 http://logs.openstack.org/77/667777/3/check/rally-tox-functional-py3/5da580a/ : SUCCESS in 5m 45s\n- rally-database-migration http://logs.openstack.org/77/667777/3/check/rally-database-migration/6507a38/ : SUCCESS in 7m 53s\n- rally-install-ubuntu-xenial http://logs.openstack.org/77/667777/3/check/rally-install-ubuntu-xenial/7b4cd16/ : SUCCESS in 3m 56s\n- rally-install-ubuntu-bionic http://logs.openstack.org/77/667777/3/check/rally-install-ubuntu-bionic/8730ed4/ : SUCCESS in 3m 22s\n- rally-install-centos-7 http://logs.openstack.org/77/667777/3/check/rally-install-centos-7/b330b71/ : SUCCESS in 5m 10s\n- rally-tox-self http://logs.openstack.org/77/667777/3/check/rally-tox-self/d2e654f/ : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"230fd86493b8eb51d94daaa98c306f51f9df4bd9","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-06-27 14:18:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1845cc4310ba421ccccdd6ff87ee86e9046e56ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:19:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"56bb49eba6b5500a735e2d0ba81765e269b923a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:40:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3e4493fe2b8de4aa49ca736c06986ffc98e42fe3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:40:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/77/667777/3/gate/openstack-tox-docs/0ef92dc/html/ : SUCCESS in 3m 27s\n- rally-tox-docs http://logs.openstack.org/77/667777/3/gate/rally-tox-docs/c6b2b6d/ : SUCCESS in 3m 43s\n- rally-tox-pep8 http://logs.openstack.org/77/667777/3/gate/rally-tox-pep8/671d4d7/ : SUCCESS in 4m 15s\n- rally-tox-py27 http://logs.openstack.org/77/667777/3/gate/rally-tox-py27/fc758d4/ : SUCCESS in 4m 58s\n- rally-tox-py34 http://logs.openstack.org/77/667777/3/gate/rally-tox-py34/4db68a6/ : SUCCESS in 4m 55s\n- rally-tox-py35 http://logs.openstack.org/77/667777/3/gate/rally-tox-py35/a8b7f43/ : SUCCESS in 5m 17s\n- rally-tox-py36 http://logs.openstack.org/77/667777/3/gate/rally-tox-py36/71c4770/ : SUCCESS in 3m 35s\n- rally-tox-py37 http://logs.openstack.org/77/667777/3/gate/rally-tox-py37/c3dee1f/ : SUCCESS in 3m 59s\n- rally-tox-cover http://logs.openstack.org/77/667777/3/gate/rally-tox-cover/1954d81/ : SUCCESS in 4m 30s\n- rally-tox-functional http://logs.openstack.org/77/667777/3/gate/rally-tox-functional/cd71a94/ : SUCCESS in 5m 14s\n- rally-tox-functional-py3 http://logs.openstack.org/77/667777/3/gate/rally-tox-functional-py3/4a7e748/ : SUCCESS in 5m 55s\n- rally-database-migration http://logs.openstack.org/77/667777/3/gate/rally-database-migration/d8aad56/ : SUCCESS in 7m 37s\n- rally-install-ubuntu-xenial http://logs.openstack.org/77/667777/3/gate/rally-install-ubuntu-xenial/2e1cdb9/ : SUCCESS in 4m 25s\n- rally-install-ubuntu-bionic http://logs.openstack.org/77/667777/3/gate/rally-install-ubuntu-bionic/2dd8101/ : SUCCESS in 3m 36s\n- rally-install-centos-7 http://logs.openstack.org/77/667777/3/gate/rally-install-centos-7/c696713/ : SUCCESS in 4m 53s\n- rally-tox-self http://logs.openstack.org/77/667777/3/gate/rally-tox-self/b32fbf5/ : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e","revisions":{"daf0abf574353b9e6260fa925890e64f758c5a3f":{"kind":"REWORK","_number":1,"created":"2019-06-27 02:56:35.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/77/667777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/77/667777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/77/667777/1"}}},"commit":{"parents":[{"commit":"6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c","subject":"Fix assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 02:50:12.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 02:50:12.000000000","tz":480},"subject":"Fix wrong assert function name","message":"Fix wrong assert function name\n\nThis is to fix wrong assert function name of asser_called_once_with\n\nChange-Id: I8ce80b7b93d4846219ad886b8531daef56e16fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/daf0abf574353b9e6260fa925890e64f758c5a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/daf0abf574353b9e6260fa925890e64f758c5a3f"}]},"branch":"refs/heads/master"},"513328c3bc0f7ff709ba0b6757319018b3fc59fa":{"kind":"REWORK","_number":2,"created":"2019-06-27 03:19:37.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/77/667777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/77/667777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/77/667777/2"}}},"commit":{"parents":[{"commit":"6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c","subject":"Fix assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 02:50:12.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 03:15:16.000000000","tz":480},"subject":"Fix wrong assert function name","message":"Fix wrong assert function name\n\nThis is to fix wrong assert function name of asser_called_once_with\n\nChange-Id: I8ce80b7b93d4846219ad886b8531daef56e16fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/513328c3bc0f7ff709ba0b6757319018b3fc59fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/513328c3bc0f7ff709ba0b6757319018b3fc59fa"}]},"branch":"refs/heads/master"},"3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e":{"kind":"REWORK","_number":3,"created":"2019-06-27 03:54:07.000000000","uploader":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"ref":"refs/changes/77/667777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/77/667777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/77/667777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/77/667777/3"}}},"commit":{"parents":[{"commit":"6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c","subject":"Fix assertion methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/6df07c6c7fc8c1cfc9acdd4643a737c7622dd52c"}]}],"author":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 02:50:12.000000000","tz":480},"committer":{"name":"zhufl","email":"zhu.fanglei@zte.com.cn","date":"2019-06-27 03:48:56.000000000","tz":480},"subject":"Fix wrong assert function name","message":"Fix wrong assert function name\n\nThis is to fix wrong assert function name of asser_called_once_with\n\nChange-Id: I8ce80b7b93d4846219ad886b8531daef56e16fa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
