)]}'
{"id":"openstack%2Frally~678330","triplet_id":"openstack%2Frally~master~I646918b8e299354e21733c090fb5ab2dec8dcf6c","project":"openstack/rally","branch":"master","topic":"docs","hashtags":[],"change_id":"I646918b8e299354e21733c090fb5ab2dec8dcf6c","subject":"Fix docs gates","status":"ABANDONED","created":"2019-08-24 06:15:09.000000000","updated":"2019-09-11 00:27:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d4f31b41d50a3ba53e3d0a175068d7f952b6034","_number":678330,"virtual_id_number":678330,"owner":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-08-24 06:41:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-24 06:41:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"686cf3775d30d25fe4808dedfd0e7f5d4f0b6414","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-08-24 06:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96a6a79b520042a114ae10634a29b6f0fb855eff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-24 06:41:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4594386125a04636b13044a98302db8d : FAILURE in 4m 46s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/cc1ac9860a8f41ff9d8c0da9c801ff3e : FAILURE in 8m 25s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/e648ae2cc5b441518c5a8b1a587e36e5 : SUCCESS in 9m 43s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/d0693ac4af5c4deaa5d2218426fcc23f : SUCCESS in 8m 48s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/71911aed79d142d699d3075b0e3a1b3d : FAILURE in 3m 59s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/a12ddef4102542a7b5e3c2823b80bbf5 : SUCCESS in 5m 19s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/dbb13712b6b74c06a2801cc99198ec15 : SUCCESS in 7m 10s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c4ca2ea353d441fc968dfd006dd903de : SUCCESS in 5m 26s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/202433b95f6c487f8d23ff2c3672712e : SUCCESS in 5m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/66af8c696d334979b20e80ca64e1955a : SUCCESS in 6m 58s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/629ae5e2a36b4be1bfe64f0287f21905 : SUCCESS in 7m 01s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/16b7ae2a5e114fb9b69e57efa8ec6842 : SUCCESS in 13m 46s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/bb51b8a7ade84c58ab8ef5672236ae59 : SUCCESS in 4m 48s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/899e5752352d416daf546342fe109143 : SUCCESS in 3m 56s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/6c49f71d246f4111adf37bc4a4ace74b : SUCCESS in 9m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/dfe95c7751db4fec9aade9ec8f0fc406 : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a259ff90e24b18dbe4c218f8c98f0bf0e9715438","author":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"date":"2019-09-11 00:27:11.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/679647/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1633f3175f8e55548c003e8c26afbf5e2cba3b8","revisions":{"b1633f3175f8e55548c003e8c26afbf5e2cba3b8":{"kind":"REWORK","_number":1,"created":"2019-08-24 06:15:09.000000000","uploader":{"_account_id":21528,"name":"chenhb","email":"chen.haibing1@zte.com.cn","username":"chenhb"},"ref":"refs/changes/30/678330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/30/678330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/678330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/678330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/30/678330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/30/678330/1"}}},"commit":{"parents":[{"commit":"3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e","subject":"Fix wrong assert function name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e"}]}],"author":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-08-24 06:53:41.000000000","tz":480},"committer":{"name":"chenhb","email":"chen.haibing1@zte.com.cn","date":"2019-08-24 06:53:41.000000000","tz":480},"subject":"Fix docs gates","message":"Fix docs gates\n\nChange-Id: I646918b8e299354e21733c090fb5ab2dec8dcf6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b1633f3175f8e55548c003e8c26afbf5e2cba3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b1633f3175f8e55548c003e8c26afbf5e2cba3b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
