)]}'
{"id":"openstack%2Frally~679047","triplet_id":"openstack%2Frally~master~I90f8512cc4a56a8092c585ecb665864a1c19b65a","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I90f8512cc4a56a8092c585ecb665864a1c19b65a","subject":"Allow plugins to return config options as list of tuples","status":"MERGED","created":"2019-08-28 13:00:14.000000000","updated":"2019-09-10 11:59:56.000000000","submitted":"2019-09-10 11:59:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679047-1568116796111-456ce0b0","meta_rev_id":"fa9fa6df46549338bf0216cf93b4f5ba5479615d","_number":679047,"virtual_id_number":679047,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-09-10 11:59:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-09-10 10:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-09-10 10:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 10:32:38.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-09-10 11:59:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14fd46a77e6e0ffea6cb1917f84431d4b6e5e93f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-08-28 13:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1775e44e058f9be9206d3e1d13077b7017db28c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:16:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936070ef464442a890cc2e08f106a2e5 : FAILURE in 4m 19s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/574e486603e044a5a17f7cf77216d7b5 : FAILURE in 4m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a716fcc35e4741876b4cced649c84d : SUCCESS in 4m 04s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/de06b51c6db04e58990b9c482870999e : SUCCESS in 3m 58s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/5f6d3ec6725a47cfb0e4866de7da21ae : SUCCESS in 7m 08s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/43b7567709ff49c3bdee059a90b5310f : SUCCESS in 5m 14s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5f662568444aa69cc2ba896276c24c : SUCCESS in 5m 59s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/46735d0606df47049408fad8b092f22a : SUCCESS in 5m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e9a1af23153d48418fbef6539070c4ab : SUCCESS in 6m 39s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/410e38d98aed48b3a0a2184200f0832f : SUCCESS in 6m 41s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/076c181e07e14afbb871a316256b63b0 : SUCCESS in 10m 15s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/8bade896655a4bd0af2a3ad90ef57ac5 : SUCCESS in 8m 18s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/8bce68e75ca24615898756a6f55213bd : SUCCESS in 5m 55s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c58b91ef91cd4c50bdd75fe89a7f0da6 : SUCCESS in 4m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/35f848d31430480a82f8623b499eb989 : SUCCESS in 7m 14s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/c6da3ae2ef0e47609f29dd1fc1082e98 : SUCCESS in 9m 21s","accounts_in_message":[],"_revision_number":1},{"id":"eceb18cb07ad94f8db20d04d0e65dfbdedccfb2f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 10:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ce08b5e10c6bac52caf0cc488692ec97758757a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:50:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ddec2718544554a725e25d3825285d : SUCCESS in 5m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/8f63266c7d9d48e6bc4f9714b0b3c555 : SUCCESS in 3m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bb9e1a35a6b4abfb500c35aaae1eab6 : SUCCESS in 4m 50s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/8c204e28092847b68ce5ea3c52ecf796 : SUCCESS in 5m 32s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/02fb50c80641498e9e5eec137e76bc27 : SUCCESS in 4m 46s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/f267cd48dff948d7bb1276a8ba8f2041 : SUCCESS in 5m 45s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/c74095056b6b414c881d3e18b46dd962 : SUCCESS in 5m 02s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/79ec8c13aabe4f2c84ba26e54b215506 : SUCCESS in 5m 27s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/78eacac56f2d43d19b7f32c7fcd38c39 : SUCCESS in 4m 19s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/89f25fa3a0fc49f2956e4370eda1a35f : FAILURE in 8m 16s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/77645d830a5a4ddaa401ffeb25ad7455 : SUCCESS in 6m 44s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/eb45ebfee1dd4c108744caf877841d06 : SUCCESS in 8m 58s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/51e550958d95428ab057d725cab73ba9 : SUCCESS in 4m 37s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77d746d9b28b496894ae41ac21971adb : SUCCESS in 3m 48s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/09a9520eba6941feaeb5a14e1cbe7823 : SUCCESS in 5m 12s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/211a66c19d574092bd412c5b08fde7ae : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a55e563b9372f9c3b08c708c6b5217ded9e1a560","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 11:20:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef3cc5ad45e538b87ee4b2f57ccd27ec386594eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:42:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b886b45d086d48be927f4c77ccc68deb : SUCCESS in 5m 38s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/19c6617afa1740a885c56662961d0a4f : SUCCESS in 3m 32s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3279497ff74892b4e952eff415d5e4 : SUCCESS in 3m 52s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/32c6a2cb3930458abdf94c7f9726d460 : SUCCESS in 3m 43s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/472169fac35d43ea9948bf6fa4ede67c : SUCCESS in 4m 57s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/230466f3a6dc489cb904412eb78b8a46 : SUCCESS in 5m 20s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/15b894dcc0744415ba49cef90d2fdbff : SUCCESS in 4m 45s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7b53b201df014e52af5973e7f97599fe : SUCCESS in 4m 45s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/c67e41e525f34e73bd24e819b4a15e54 : SUCCESS in 5m 18s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/cf0c40dc7b184fbc8adab8194524fa18 : SUCCESS in 6m 22s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/129da87fe42a4b93bd57879769fb4015 : SUCCESS in 7m 41s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/77d9cdba5fcc4221a5b25504a87333b5 : SUCCESS in 8m 21s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/47ada9b5741643138fd3ed2b4536fb8c : SUCCESS in 3m 57s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2dc55b2e5b7945eda6fb045c0a410e4e : SUCCESS in 3m 32s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5d49b6192a494716b06f70007799098a : SUCCESS in 5m 19s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/35e562bca2f641a0ad563a40016d9c3c : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2bf85a31862b50dac4aa41597435ad1d89280e50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:42:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fbf4dd481f574752b60e56c25ce775580942ba7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:59:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc9aa5361ea84766a629a3d8c76c0b7c : SUCCESS in 3m 50s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/b2e55a493446432f8ead8516d83e55eb : SUCCESS in 3m 24s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/548b40d1ae0945a3a35a6de15a193161 : SUCCESS in 3m 41s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/ef4c7532f8184941b0922f9ac0fbcb8e : SUCCESS in 3m 51s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/5d2691003952498f885a1f4b1d8c64e2 : SUCCESS in 5m 41s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/2403978ea9334e5a8b546d4cff537cb6 : SUCCESS in 5m 51s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b8c78101c849c5a066a92fe7123a2e : SUCCESS in 3m 45s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4181c8973cd1420dbd3eb21b676de457 : SUCCESS in 4m 18s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e824d62218f04cc8a599bd27a4a37bd4 : SUCCESS in 6m 01s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/dae47f6486e3492094efd1b677012140 : SUCCESS in 6m 33s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/aa2db5159a7c4e6183bd055a4307b438 : SUCCESS in 5m 46s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3c0507804f164a3fb30e1b83e02889ee : SUCCESS in 9m 39s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/693629ab325f4308a4b45f0509793b21 : SUCCESS in 4m 14s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88307bfd349a4f17b3f39e7e33d8c12d : SUCCESS in 3m 34s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/d694ee7bc93b4b2da980bfcabd8a7fc0 : SUCCESS in 4m 54s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9673ac6503e6423482286020449d1c69 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fa9fa6df46549338bf0216cf93b4f5ba5479615d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:59:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54c3d83200b21f3cc6509ca4637d72d213d77dd5","revisions":{"54c3d83200b21f3cc6509ca4637d72d213d77dd5":{"kind":"REWORK","_number":1,"created":"2019-08-28 13:00:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/47/679047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/47/679047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/679047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/679047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/47/679047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/47/679047/1"}}},"commit":{"parents":[{"commit":"3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e","subject":"Fix wrong assert function name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3c529c7c4f666488e3eadb6a4dd06b881e0ebb3e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-08-28 12:56:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-08-28 12:56:16.000000000","tz":60},"subject":"Allow plugins to return config options as list of tuples","message":"Allow plugins to return config options as list of tuples\n\noslo-config-generator expects the config options to be a list of tuples:\n\nhttps://docs.openstack.org/oslo.config/latest/cli/generator.html\n\nThis change allows us to us the same format and avoid having to define\ntwo virtually identical functions.\n\nChange-Id: I90f8512cc4a56a8092c585ecb665864a1c19b65a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/54c3d83200b21f3cc6509ca4637d72d213d77dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/54c3d83200b21f3cc6509ca4637d72d213d77dd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
