)]}'
{"id":"openstack%2Frally~681224","triplet_id":"openstack%2Frally~master~Icf5651117b20a8ecfe6a3e5f42eb3a4e569e0618","project":"openstack/rally","branch":"master","topic":"rally_1.7.0","hashtags":[],"change_id":"Icf5651117b20a8ecfe6a3e5f42eb3a4e569e0618","subject":"Remove a bunch of deprecated stuff from CLI","status":"MERGED","created":"2019-09-10 11:33:23.000000000","updated":"2019-09-10 14:21:08.000000000","submitted":"2019-09-10 14:21:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681224-1568125268452-d42ba417","meta_rev_id":"717aa095016307a06bf9971b4ed4109dca9da698","_number":681224,"virtual_id_number":681224,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-09-10 14:21:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-09-10 13:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-09-10 13:04:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 14:21:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73cbcb6b8dad07f66f66af1ca11ac8517ccf4a5e","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 11:33:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4df2fef54decfa57305ab27990d126b861b0771e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:48:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ca6d383ce9b4ccb9317c5ff07973b7b : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34927dcb83a94b75b89da27486c3859f : SUCCESS in 5m 07s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/36084c899ee640d78f7ea50db49646ed : SUCCESS in 3m 42s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e573ef923c404980a979a3df689ac3 : SUCCESS in 4m 33s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/56026c9eb84741ba9f0582c5924fa08b : SUCCESS in 3m 54s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/91dd4a9d10504ea986a008118bb59b95 : SUCCESS in 5m 10s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/c245a9428e724864893bcfc470a56387 : SUCCESS in 4m 37s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2e9aa18207ef45c29b48325f741dd32e : SUCCESS in 5m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f0b82dd14cfd43e1801eff7fd6f8ad07 : SUCCESS in 4m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2da59944d8c440b7949d8976bfa12ce2 : FAILURE in 4m 50s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a8ae4f709ad34d5fac883eb26ea4a986 : FAILURE in 6m 16s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/0fa921a3cdc4407d8fef753dfe720fdc : FAILURE in 6m 39s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/9d6fdfbd72194e11b61fac342348d074 : SUCCESS in 7m 35s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c3cd4068062b4b49b94039bb1d8cd07d : SUCCESS in 4m 26s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cda73342f7d142dd93d82df9c7d7882c : SUCCESS in 11m 33s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/edff76192344496a986bf143c104cf62 : SUCCESS in 5m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/71f2fb81f3934b6799af5c2d89ebbb67 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"59e700b6e7abce31adf68efb07281b31bfbd4683","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 12:01:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c6b20c2083bba7f3f4890f6ece3c893543d2e643","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:15:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6793551c56bd4c25b1606e3ccc63aa8d : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e4b554902d4d569954209aedb17acd : SUCCESS in 4m 22s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/ddf91225bf2744e7a7ad982e19c05f78 : SUCCESS in 4m 22s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5dcc7b6fd146ddb35af6f94c49edb2 : FAILURE in 4m 00s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/b448c8ce6cd14e1babffd7c711e30b56 : SUCCESS in 3m 43s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/81afaa2d2ee04a73a91aff3875f22df7 : SUCCESS in 6m 52s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/8dab74a0ae504ff390ebc9f1abf4b828 : SUCCESS in 5m 22s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/69c2422a9bc94c58b2a3c88410dd6935 : SUCCESS in 4m 06s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/60596443e5a948e3924d82bba01abddc : SUCCESS in 4m 23s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/adb5ca8a75ca4f1698e21435465f6026 : SUCCESS in 4m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/7ac842253d2b48cab701415e4825b051 : FAILURE in 5m 39s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/16d4b9a31dd548cca3170947677ff7ca : FAILURE in 6m 55s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/069fa2cfbb9c4354b74fc9b7ef403a56 : SUCCESS in 7m 31s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9e117e25e19a4ab4a0c3be29920733e0 : SUCCESS in 3m 34s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c04b9b86544c40e78d37455bf5e6487c : SUCCESS in 3m 55s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/41cebce548bd489ea73cf420521931af : SUCCESS in 4m 48s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/41d9d654585b46ac91d29b7fd47c41b0 : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3d470013f66dc4524663adc2adcde8f2bc803f73","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 12:49:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1667931fcde362acdb9a8dcbfc438245991a2dc1","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 13:04:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"af5732478ae2734e6955aa07d82ad8c46f710de5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:07:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/970d6976602841f9bb949bba4703e2c0 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdde8e28307a4f2f9dfa0e4bdaf1cddb : SUCCESS in 4m 26s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f9a57c562dba449ca0a60dfe23241a31 : SUCCESS in 4m 12s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d505d0d82460415797f409b39d39d267 : SUCCESS in 4m 23s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/b0668bad8e0d4ea2be46720c149c0245 : SUCCESS in 3m 42s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/da6b351af4d04930ba97fed9320b5589 : SUCCESS in 4m 41s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/6650506b6c434c12ae57feaa92f64bdd : SUCCESS in 4m 43s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/ba6ddf0706424646ac35d9b19fd56aca : SUCCESS in 4m 00s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4d3218edb68c4e95a8c318e4866f8d00 : SUCCESS in 4m 13s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0ab18d57f89449bbb10db81e6928ac5b : SUCCESS in 5m 45s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/95f622cbf2c5473c86f0a79952440079 : SUCCESS in 5m 32s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/5761c9c50a5c4723af14cae6ef3317f3 : SUCCESS in 5m 55s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/6c697f2da180406290346d4042242a46 : SUCCESS in 7m 44s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f44b84b8e1b24dbba0fe4f3306636802 : SUCCESS in 4m 07s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07603891641a43e09f897050180c9fda : SUCCESS in 3m 31s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/ee10fe973fbd4c0f9f6ee0a3257963f3 : SUCCESS in 4m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/923b309463e142c49b59186b3b705fe7 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"07ea448069dbee463d43d30ac2dd6f7a66513829","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:07:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b3d68c08b202e5d42d21bfad2185b2dd3d99fdb5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:19:08.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51a4eb8092384b508120c0c91d46c81a : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99dfa3711e99417fa3ff4ed4454e08bd : SUCCESS in 4m 26s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/dda2fbb5e38643aa870ee00873a3582c : SUCCESS in 3m 27s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed1a55f16c04d7791027d985e3bac29 : SUCCESS in 3m 48s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/7fada702ece644b4a8e5aad3ffd9f38b : SUCCESS in 3m 48s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/9787ac807f0b45d1a921dff6137f3b50 : SUCCESS in 5m 09s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/02998ed72f1f42de97c08d43d53e0e5a : FAILURE in 6m 06s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/6abf27b49ac343359b7753b51cc2902e : SUCCESS in 4m 32s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/944cf7488ef34924932fda8eb98a0e2e : SUCCESS in 5m 28s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/0520220e1b954424a2c4befdca245b46 : SUCCESS in 5m 37s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1a049873ec6d4cbf98d1e5882dd9de88 : SUCCESS in 6m 11s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/1bf4690116194e9ebc65f187a25e5d0e : SUCCESS in 7m 13s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/a4d954960c234db8bba8699ecfc164b3 : SUCCESS in 8m 09s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d445c8280b294e8da972c7cffd6ca1ad : SUCCESS in 3m 46s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da2d3b96c4004d06931764b83e63d866 : SUCCESS in 3m 29s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/24c2f8799f1e4524be43166a4613eda6 : SUCCESS in 5m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/9941e19fbd2943e283913769170318d4 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4161706ee79a9f11bd8302c8debe4d81e380335f","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 13:48:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"05801955f29cd8cb2b39eee17cc8ed26c30de0b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:05:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8b6990430014764953aa401fcff1c82 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40927d300a0c42c5915281fed1205496 : SUCCESS in 5m 55s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d7d746becae64b7f8441a63a59852bc5 : SUCCESS in 4m 50s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f384c388d80463bbfe16c0f9f492a19 : SUCCESS in 5m 24s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/bffb2343b3e641959fd2c82a75f6ee3f : SUCCESS in 8m 14s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/16781aa6b43e4560885a284c84e99733 : SUCCESS in 7m 22s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/c3f757a7a9d34320abf1739e1758d3d9 : SUCCESS in 4m 29s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/75a6fa68408a4f739200d4301cb94d90 : SUCCESS in 4m 40s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a75eab9629344598a2ad99444d939943 : SUCCESS in 4m 47s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/a8d89e2b885d4bc5a01ccd17787879a7 : SUCCESS in 6m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/6b2b58c10a3648599aee186b80486b3d : SUCCESS in 5m 28s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/ab6e74163a354bd7b1fc3e0809d285b4 : SUCCESS in 6m 04s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/3bb1b2d9c41945b7bd9ceb930e021c0c : SUCCESS in 8m 46s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/a84648c3ce2b4f888ba47fe71f0bbea8 : SUCCESS in 5m 13s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3d398c79f5e42bcad67db4c54415ca5 : SUCCESS in 4m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/0da55c2d0e8c4d428da54949fb494adf : SUCCESS in 5m 18s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/07c60a21d9b64b90aaa4b4b7452010e6 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f822df9065e7efe0629cd69223597e237d5e2dae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:05:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a0b4c628f0c3f933879e9547ec0139a8b43a94cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:21:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f41541fb2b9dc764dfff598e6bfddc18c01f0817","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:21:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed661efedf2c49f4aafd9a257e88dd19 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f52e67d6754bc1bde9586a7daf470b : SUCCESS in 4m 06s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/062e72b09cd0460cb0f309d5f63fa2f8 : SUCCESS in 4m 34s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/8770265adec443ed838c10f12f816d9a : SUCCESS in 4m 03s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/d094826b33ac439fb35dcbec8a4bf617 : SUCCESS in 4m 26s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/3a87e9d879344e9f9d0f2802ae6bc607 : SUCCESS in 4m 30s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/5d358333d16943cab49ce655f9ab83c1 : SUCCESS in 4m 29s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/7feebf5e2f3d407697109baa625ca7e7 : SUCCESS in 5m 12s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/76fb372e1f144f8a844b6025dfd05efb : SUCCESS in 5m 48s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/bb641dbc112b43beb9cb7cc11a62fd2d : SUCCESS in 5m 05s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/44ff4b9e35604f2aa557c22b773f6984 : SUCCESS in 6m 14s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/1dd66ce1f1d3439cb6984b45b4fbe98f : SUCCESS in 5m 38s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/d12c2f718d2848a8962a4ed6a1d588fd : SUCCESS in 7m 32s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/59b747ff8b3d46e898f1570134ff32f4 : SUCCESS in 3m 38s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a28fa682544a4a2db6f66124369d17f2 : SUCCESS in 3m 28s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5608cbd1a9ea43bda890ad711db628e6 : SUCCESS in 6m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0d453a35b227413fb8edde161e72bc5e : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8d54b0e235da0c853513b9df06c51f6e86116f4b","revisions":{"4e23629c2d223c88168ce725f498a14ae26c31de":{"kind":"REWORK","_number":1,"created":"2019-09-10 11:33:23.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/24/681224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/24/681224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/24/681224/1"}}},"commit":{"parents":[{"commit":"60d39f3a37b46a32afed3a83b5a6346c4c230b43","subject":"Upgrade jsonschema upper version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/60d39f3a37b46a32afed3a83b5a6346c4c230b43"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 11:27:19.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 11:27:19.000000000","tz":180},"subject":"Remove a bunch of deprecated stuff from CLI","message":"Remove a bunch of deprecated stuff from CLI\n\n* rally-manage command is deprecated since rally 0.10.0\n* `--namespace` argument is deprecated since rally 0.10.0\n\nChange-Id: Icf5651117b20a8ecfe6a3e5f42eb3a4e569e0618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e23629c2d223c88168ce725f498a14ae26c31de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/4e23629c2d223c88168ce725f498a14ae26c31de"}]},"branch":"refs/heads/master"},"b44f02ca7b429e457f33051f91023734993767e0":{"kind":"REWORK","_number":2,"created":"2019-09-10 12:01:10.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/24/681224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/24/681224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/24/681224/2"}}},"commit":{"parents":[{"commit":"60d39f3a37b46a32afed3a83b5a6346c4c230b43","subject":"Upgrade jsonschema upper version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/60d39f3a37b46a32afed3a83b5a6346c4c230b43"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 11:27:19.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 12:00:34.000000000","tz":180},"subject":"Remove a bunch of deprecated stuff from CLI","message":"Remove a bunch of deprecated stuff from CLI\n\n* *rally task sla_check* command was deprecated in Rally 0.8.0 in favor of\n  *rally task sla-check*.\n\n* *rally-manage db* command (and the whole *rally-manage* entry-point) was\n  deprecated in Rally 0.10.0 in favor of *rally db* command.\n\n* *--namespace* argument was deprecated in Rally 0.10.0 in favor of\n  *--platform* which has better meaning.\n  Affected commands: *rally plugin show*, *rally plugin list*,\n  *rally verify list-plugins*, *rally verify create-verifier*.\n\n* *--tasks* argument of *rally task report* command and *--task* argument of\n  *rally task use* command were deprecated in Rally 0.10.0 in favor of\n  unified *--uuid* argument.\n\n* *--junit* argument of *rally task report* command is deprecated in\n  Rally 0.10.0 in favor of *rally task export --type junit-xml*\n\nAlso, this patch covers @deprecated_args cli decorator to pass cover\njob.\n\nChange-Id: Icf5651117b20a8ecfe6a3e5f42eb3a4e569e0618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b44f02ca7b429e457f33051f91023734993767e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b44f02ca7b429e457f33051f91023734993767e0"}]},"branch":"refs/heads/master"},"8d54b0e235da0c853513b9df06c51f6e86116f4b":{"kind":"REWORK","_number":3,"created":"2019-09-10 12:49:45.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/24/681224/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/24/681224/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/24/681224/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/24/681224/3"}}},"commit":{"parents":[{"commit":"60d39f3a37b46a32afed3a83b5a6346c4c230b43","subject":"Upgrade jsonschema upper version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/60d39f3a37b46a32afed3a83b5a6346c4c230b43"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 11:27:19.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 12:49:38.000000000","tz":180},"subject":"Remove a bunch of deprecated stuff from CLI","message":"Remove a bunch of deprecated stuff from CLI\n\n* *rally task sla_check* command was deprecated in Rally 0.8.0 in favor of\n  *rally task sla-check*.\n\n* *rally-manage db* command (and the whole *rally-manage* entry-point) was\n  deprecated in Rally 0.10.0 in favor of *rally db* command.\n\n* *--namespace* argument was deprecated in Rally 0.10.0 in favor of\n  *--platform* which has better meaning.\n  Affected commands: *rally plugin show*, *rally plugin list*,\n  *rally verify list-plugins*, *rally verify create-verifier*.\n\n* *--tasks* argument of *rally task report* command and *--task* argument of\n  *rally task use* command were deprecated in Rally 0.10.0 in favor of\n  unified *--uuid* argument.\n\n* *--junit* argument of *rally task report* command is deprecated in\n  Rally 0.10.0 in favor of *rally task export --type junit-xml*\n\nAlso, this patch:\n\n* covers @deprecated_args cli decorator with unit teststo pass cover job.\n\n* switches functional tests to use testtools library instead of unittest\n (for unificaiton with our unit tests)\n\nChange-Id: Icf5651117b20a8ecfe6a3e5f42eb3a4e569e0618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d54b0e235da0c853513b9df06c51f6e86116f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d54b0e235da0c853513b9df06c51f6e86116f4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
