)]}'
{"id":"openstack%2Frally~681234","triplet_id":"openstack%2Frally~master~I52b9ff6458675d402295e8f0bde576afe2c53f37","project":"openstack/rally","branch":"master","topic":"rally_1.7.0","hashtags":[],"change_id":"I52b9ff6458675d402295e8f0bde576afe2c53f37","subject":"Make rally 1.7.0","status":"MERGED","created":"2019-09-10 12:06:01.000000000","updated":"2019-09-10 14:24:09.000000000","submitted":"2019-09-10 14:24:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681234-1568125449056-caf46e47","meta_rev_id":"f2bd38547f198ed2a3e14933e1e19f5fdf492f9e","_number":681234,"virtual_id_number":681234,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":2,"date":"2019-09-10 14:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-09-10 13:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-09-10 13:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 14:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b4ede48901ea7439e8373861cdf94391f2dc0b7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 12:06:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8aff846cecb71bd3893762b0121afb1f281c45e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:22:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f70197e7b344a93a54fa3003edd8686 : SUCCESS in 4m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/a6efa18224b44de79972d953b9a79354 : SUCCESS in 6m 11s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/431e4752945e43bebf3cebaeee48d12b : FAILURE in 5m 12s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/2291ba149d3149e29450e0341cff5b38 : SUCCESS in 5m 15s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/7d6d2f0513264617bebf0b01225c493a : SUCCESS in 7m 47s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/d67c728f309f4ad6828f83f2bd65a83c : SUCCESS in 4m 44s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/9de78cec4874439cb4315b835d5a3f48 : SUCCESS in 3m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/dc9e78a959274f73a1d96d1e89dc6155 : SUCCESS in 5m 13s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/baba9add3de04aa29aa7253fb6c405b2 : SUCCESS in 5m 20s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/557c09c5904b45c58ac6f1359fdd5c7a : FAILURE in 6m 02s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/ba821c681910449683991e8861121a1e : FAILURE in 8m 38s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/721e3600d5bc48ccbcbeda0b8220aaf7 : SUCCESS in 7m 39s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f57256d643fa4046bb13fd6f70da2e44 : SUCCESS in 6m 04s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/043958a98d7c496c9204a55f678374e7 : SUCCESS in 5m 48s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/b0789b2ce93a44f8acc63166359210c0 : SUCCESS in 7m 53s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/2b832aa13dfc4ea6a26b2f8a2a491888 : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fadf9813a923b6f807205fd674ec2e384070f7b7","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 13:04:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"873f94f0384afb9a00acc01b87971028fb8d4f04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:24:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/468b17e4a99241f6a2b1f751aaea2d8c : SUCCESS in 3m 53s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/637a486538a34e2b902222e5dc2fe409 : SUCCESS in 3m 25s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb5f29f1bc841279f2ceb244ed56e9d : SUCCESS in 4m 15s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/62e47411c8744746a18b4dca8c36f658 : SUCCESS in 4m 21s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/eba5596d008143d38bc93ac0aea4562d : SUCCESS in 6m 44s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/0020f9f16bb84450a432d625d150bed4 : SUCCESS in 4m 00s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/89b834dbe67f461ab0b93571c268787b : SUCCESS in 4m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/12f54efd6a8746ee8ebea4fffac4d29b : SUCCESS in 5m 46s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/89e7b09dc64b4c69ba250ec5f9531446 : SUCCESS in 4m 56s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/92987b0285eb4180ba5ee21eb3039092 : SUCCESS in 6m 06s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/8e10434a25964a37bae062d80813d802 : SUCCESS in 6m 15s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/8e28f712ba004ffd97761a175ba5edfb : SUCCESS in 8m 22s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fe6ec99d4a1f4779bca5fba8b1360cd5 : SUCCESS in 5m 29s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25afbbbf5c444fa98d290eec68f2249a : SUCCESS in 4m 20s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e5ed6b89b87c4b578a60243aa52c4f9c : SUCCESS in 4m 46s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/330169d7ca70471aa58d7cb860a93b89 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":2},{"id":"eada14a7483b67367624623226964949070c83cb","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-09-10 13:27:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0aef7d945342e377d0864190b436d0065c6654ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:05:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8da3c56b6c32d8556e066cc6526223442d89bc15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:24:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec6da9978b043ddb94b53c6772da4b8 : SUCCESS in 4m 11s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/455f86aef9034f2b8266f6bd49e1f906 : SUCCESS in 4m 16s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0c9a3a7f9c42c2b3df2db1894755db : SUCCESS in 4m 24s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/68e66f564c884d6c81ece521246d63b8 : SUCCESS in 4m 28s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/c8e4f3481a574b54ae2838884c6ebaf1 : SUCCESS in 4m 36s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/be75710655d74f4eb823cc4976dc5c9d : SUCCESS in 5m 29s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bc709282f14a21ba158a1b65a359c6 : SUCCESS in 3m 50s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f35405382d8b4eb5aac77b5785174992 : SUCCESS in 5m 53s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/431eb7deb789483fa70b5175cee6c3cd : SUCCESS in 5m 14s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a16e6cb7128643c7807e4ad472dd6558 : SUCCESS in 7m 07s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/d051a93192cb438092afbb59a221086d : SUCCESS in 5m 40s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/ba36b2c17f1442b68714843f3d1d6c1c : SUCCESS in 8m 35s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f0fd6a4a8f004a5086f21077bde91667 : SUCCESS in 3m 48s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/006714e8502d48b2b1f7f03b2bf8af6f : SUCCESS in 4m 19s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/1c576e04a5ce4ad184c10cb4f1625d32 : SUCCESS in 5m 16s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/e66184d06ba348fc9bb2fe09ce685f27 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":2},{"id":"cfc94cd700931e8e39e5c8c1348c82b574fee4c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 14:24:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd939c5913f15554b8aab428ea0f6106ad51af07","revisions":{"1541c7d2562fdfe84c789cf389a4a32ffd95c9fb":{"kind":"REWORK","_number":1,"created":"2019-09-10 12:06:01.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/34/681234/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/34/681234/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/34/681234/1"}}},"commit":{"parents":[{"commit":"b44f02ca7b429e457f33051f91023734993767e0","subject":"Remove a bunch of deprecated stuff from CLI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b44f02ca7b429e457f33051f91023734993767e0"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 12:04:45.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 12:05:53.000000000","tz":180},"subject":"Make rally 1.7.0","message":"Make rally 1.7.0\n\nChange-Id: I52b9ff6458675d402295e8f0bde576afe2c53f37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1541c7d2562fdfe84c789cf389a4a32ffd95c9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1541c7d2562fdfe84c789cf389a4a32ffd95c9fb"}]},"branch":"refs/heads/master"},"bd939c5913f15554b8aab428ea0f6106ad51af07":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-10 13:04:37.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/34/681234/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/34/681234/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/34/681234/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/34/681234/2"}}},"commit":{"parents":[{"commit":"8d54b0e235da0c853513b9df06c51f6e86116f4b","subject":"Remove a bunch of deprecated stuff from CLI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/8d54b0e235da0c853513b9df06c51f6e86116f4b"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 12:04:45.000000000","tz":180},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-09-10 13:04:37.000000000","tz":0},"subject":"Make rally 1.7.0","message":"Make rally 1.7.0\n\nChange-Id: I52b9ff6458675d402295e8f0bde576afe2c53f37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bd939c5913f15554b8aab428ea0f6106ad51af07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bd939c5913f15554b8aab428ea0f6106ad51af07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
