)]}'
{"id":"openstack%2Frally~695557","triplet_id":"openstack%2Frally~master~Ia5735f4d8cf54e6e7cd449947ce4db9b38bbbd52","project":"openstack/rally","branch":"master","topic":"bug/1853505","hashtags":[],"change_id":"Ia5735f4d8cf54e6e7cd449947ce4db9b38bbbd52","subject":"Fix https_key test","status":"MERGED","created":"2019-11-21 19:32:09.000000000","updated":"2019-11-26 19:07:18.000000000","submitted":"2019-11-26 19:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695557-1574795238625-de65efff","meta_rev_id":"1d9fbdf4535030f03c33f974fd3d5d552cea5301","_number":695557,"virtual_id_number":695557,"owner":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 19:07:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-11-26 18:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-11-26 18:52:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 18:52:05.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-11-26 19:07:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a069b5fe1de938e3a4acc77ca50e4487bfdc0b1d","author":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"date":"2019-11-21 19:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"869b06ee618bb70fb385aa62283f7b34719eb7fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 19:53:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306627a7cd6b4555999c08dd260953a0 : SUCCESS in 5m 09s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/6a30c8dcad674f22b9929f4b2b7263ff : SUCCESS in 3m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/5358282ea89c4b729630e876e396d368 : FAILURE in 4m 03s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/142f1b3f45b44339b9a06fb7e506bf7b : SUCCESS in 3m 59s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/f8bc638e9e7647259511c91f3cbe0289 : SUCCESS in 4m 42s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/626589a7c60841a7aceda58f630f9543 : SUCCESS in 4m 25s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/cfc6489e30604a7290f30a6ca3a3a08f : SUCCESS in 3m 48s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/6310d5be4b40431f99d86b4d253cbde5 : SUCCESS in 4m 24s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/f7ba6dff24cb4e9aa4495c407eeab084 : SUCCESS in 4m 49s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/312186d4a2ed42f0b812eee09de75e12 : SUCCESS in 8m 24s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/760b41f819a5484bb1c4f5e35aaadf7e : SUCCESS in 7m 27s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/2b39080048a748e7be5b46e4d44b2a0a : SUCCESS in 8m 36s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b084c02250fa4ee9ae1ea2027e5bcbc3 : SUCCESS in 5m 57s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cbe11e5e188414481eef3c9c3e9e502 : SUCCESS in 4m 53s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/f6d016421e8d4bbeb1e2d9b48bf3bd58 : SUCCESS in 5m 47s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/71b942eb4657405da212fce699aeb5b1 : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5bce41f5aa06cb64883df6623b928595d202df95","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-25 16:06:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ad11bc7fd1ba8d8ebda6f02f728bc76b2244daeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 16:19:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33d5bcbf1e3416fbae271ccecdcaba7 : SUCCESS in 5m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/77049ecb5c074c0db5eb79b313a7b329 : SUCCESS in 5m 49s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c69bc32d8fd404986911b1d947bbca4 : SUCCESS in 4m 52s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/100a18da35db4c1ebbdfb877bf3430a2 : SUCCESS in 5m 42s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/2b0a235dd0b4496389336e29d2a6fd25 : SUCCESS in 4m 40s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/eab7faab05814950954c7f3153a01c1e : SUCCESS in 5m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/641ba752dcf345ffaf76cc9b6e97c3df : SUCCESS in 6m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/caf3b033f848431bb3441b9561c7e3be : SUCCESS in 4m 11s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/e58763d22d1145e3a1bc1f24a93b9bed : SUCCESS in 5m 57s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/90da8b434cdc420db08e7b75546100ab : SUCCESS in 8m 40s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/aac417ebfee1424aa34e456be41ad326 : SUCCESS in 6m 58s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/a9f386203aee4d699272ad56f87a4847 : SUCCESS in 7m 43s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9afd1a03186541d9ad1f827dc7318051 : SUCCESS in 4m 00s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b36e30216d7482aa41b5a67dc74301b : SUCCESS in 4m 25s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/048fdb02656c4c978533774c77df8743 : SUCCESS in 6m 04s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/a8c8f6f497f9444ebd3e745b952a238c : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":2},{"id":"06bad942490c68fe77d5566d5a20fb939a6e9638","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-26 18:52:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"24a59ea1fabda26b64a6cd2ee331a1ef463d2e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:52:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1797a2e77471469bd91dbfd5ffecd3a3b5941180","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:07:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"063f0b44d35590ec1e48dae1626e31edf5665747","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:07:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4657674a569d477298373832db8d7d92 : SUCCESS in 4m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/7410ca1bfbbe4f73accdc2cb7542dfb4 : SUCCESS in 3m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/d32c8b8235804f27bb063cfac00b9a1e : SUCCESS in 5m 22s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/f3904e1e0da942a4aa54249ead20f42e : SUCCESS in 3m 36s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/8aa710207d5a4fa199a2be813b85cb10 : SUCCESS in 5m 06s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/25cde3fa5ccd4c038e6b0dbf28cae03f : SUCCESS in 5m 30s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6cdeb6234f4813b14d66a5c115c5da : SUCCESS in 4m 04s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/f5c84f9b22c14b2ab5c23704a33f34ba : SUCCESS in 4m 32s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/96547d0a284c44eabd77526f47dcb515 : SUCCESS in 4m 19s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/da014a5a39b44e8eb54d33bbfaa0216c : SUCCESS in 6m 31s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/cc7f5abfdd3a4319a573f92eb134947d : SUCCESS in 6m 29s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/af5b076cb6e44fab9bf5c02e0e7d215b : SUCCESS in 7m 19s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0405d636f1624f7090192c3e7d438739 : SUCCESS in 4m 39s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/825341e52d094da7b8c2c783b71bd609 : SUCCESS in 3m 51s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/04ba331f9c284e86abb09eaa9e05a0de : SUCCESS in 5m 05s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/0bc7ff893b9b4410953fd56d553c6572 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"996b060601b5f7e7221aa752da464df08a687d25","revisions":{"ecdcc3c90bad8b79dda36a662a9b36aca7da633e":{"kind":"REWORK","_number":1,"created":"2019-11-21 19:32:09.000000000","uploader":{"_account_id":27941,"name":"Jeff Albert","email":"jralbert@uvic.ca","username":"jralbert"},"ref":"refs/changes/57/695557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/57/695557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/57/695557/1"}}},"commit":{"parents":[{"commit":"a2cc6e41ffbf680c455893be63ab54230be52728","subject":"Merge \"Remove redudant \u0027be\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a2cc6e41ffbf680c455893be63ab54230be52728"}]}],"author":{"name":"Jeff Albert","email":"jralbert@uvic.ca","date":"2019-11-21 19:29:54.000000000","tz":-480},"committer":{"name":"Jeff Albert","email":"jralbert@uvic.ca","date":"2019-11-21 19:31:40.000000000","tz":-480},"subject":"Fix https_key test","message":"Fix https_key test\n\nTests for existence of https_key dictionary key before attempting to\ncheck its value; moves the pop command into the if block so it\u0027s\ngoverned by that test.\n\nChange-Id: Ia5735f4d8cf54e6e7cd449947ce4db9b38bbbd52\nCloses-Bug: #1853505\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ecdcc3c90bad8b79dda36a662a9b36aca7da633e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/ecdcc3c90bad8b79dda36a662a9b36aca7da633e"}]},"branch":"refs/heads/master"},"996b060601b5f7e7221aa752da464df08a687d25":{"kind":"REWORK","_number":2,"created":"2019-11-25 16:06:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/57/695557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/57/695557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/57/695557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/57/695557/2"}}},"commit":{"parents":[{"commit":"a2cc6e41ffbf680c455893be63ab54230be52728","subject":"Merge \"Remove redudant \u0027be\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/a2cc6e41ffbf680c455893be63ab54230be52728"}]}],"author":{"name":"Jeff Albert","email":"jralbert@uvic.ca","date":"2019-11-21 19:29:54.000000000","tz":-480},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-25 16:06:02.000000000","tz":0},"subject":"Fix https_key test","message":"Fix https_key test\n\nTests for existence of https_key dictionary key before attempting to\ncheck its value; moves the pop command into the if block so it\u0027s\ngoverned by that test.\n\nChange-Id: Ia5735f4d8cf54e6e7cd449947ce4db9b38bbbd52\nCloses-Bug: #1853505\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
