)]}'
{"id":"openstack%2Frally~696537","triplet_id":"openstack%2Frally~master~Ic94ab503ea77c8a1582292becad872670034d49c","project":"openstack/rally","branch":"master","topic":"py3","hashtags":[],"change_id":"Ic94ab503ea77c8a1582292becad872670034d49c","subject":"Stop supporting python \u003c3.6","status":"MERGED","created":"2019-11-28 13:00:14.000000000","updated":"2020-02-10 14:39:41.000000000","submitted":"2020-02-10 14:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696537-1581345581193-9ea6cd4b","meta_rev_id":"50b227d2d6aa896ec59a0941fb119050196f335d","_number":696537,"virtual_id_number":696537,"owner":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-02-10 14:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-02-10 14:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 14:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6181efee6d5bebfeadfdacccd41f63433ff9cc09","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-28 13:00:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a80fa41034faee863ddaad8d91fce5c115fd895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:00:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/rally-openstack on branch master.  The error was:\n\n  Job rally-tox-py27 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-pep8\n          - rally-tox-py27\n          - rally-tox-py35\n          - rally-tox-py36\n          - rally-tox-py37\n          - rally-dsvm-tox-functional\n          - rally-docker-check\n          - rally-task-basic-with-existing-users:\n              # use_existing_users key did not trigger proper ansible tasks\n              voting: false\n          - rally-task-simple-job\n          - rally-task-barbican:\n              voting: false\n          - rally-task-cinder\n          # NOTE(andreykurilin): this requires more thing to configure before\n          #   launching.\n          #- rally-task-designate\n          - rally-task-heat:\n              voting: false\n          - rally-task-ironic\n          - rally-task-keystone-glance-swift\n          - rally-task-magnum:\n              voting: false\n              files:\n                - rally-jobs/magnum.yaml\n                - rally_openstack/cleanup/resources.py\n                - rally_openstack/contexts/magnum\n                - rally_openstack/scenarios/magnum\n                - rally_openstack/osclients.py\n          - rally-task-manila-no-ss:\n              voting: false\n          - rally-task-manila-ss:\n              voting: false\n          - rally-task-mistral\n          # it did not work for a long time. try to re-configure it\n          #- rally-task-monasca\n          - rally-task-murano:\n              voting: false\n          - rally-task-neutron\n          - rally-task-neutron-trunk:\n              files:\n                - rally-jobs/neutron-trunk.yaml\n                - rally_openstack/scenarios/neutron/trunk.py\n                - rally_openstack/scenarios/neutron/network.py\n          - rally-task-neutron-with-extensions:\n              voting: false\n          - rally-task-nova:\n              voting: false\n          # it did not work for a long time\n          #- rally-task-senlin\n          #- rally-task-octavia:\n          #    voting: false\n          - rally-task-telemetry\n          - rally-task-watcher:\n              # watcher-api did not start last time\n              voting: false\n          - rally-task-zaqar\n          - rally-verify-tempest\n      gate:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-pep8\n          - rally-tox-py27\n          - rally-tox-py35\n          - rally-tox-py36\n          - rally-tox-py37\n          - rally-dsvm-tox-functional\n          - rally-docker-check\n          - rally-task-simple-job\n          - rally-task-cinder\n          #- rally-task-heat\n          - rally-task-ironic\n          - rally-task-keystone-glance-swift\n          - rally-task-mistral\n          - rally-task-neutron\n          - rally-task-telemetry\n          - rally-task-zaqar\n          - rally-verify-tempest\n\n  in \"openstack/rally-openstack/.zuul.d/zuul.yaml@master\", line 12, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0066788c9e5bdd0003adede4fe51fd13395c555a","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-28 13:03:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"daaac41ceea6ec13f2217db45681193cbadbdc98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:03:23.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/rally-openstack on branch master.  The error was:\n\n  Job rally-tox-py27 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-pep8\n          - rally-tox-py27\n          - rally-tox-py35\n          - rally-tox-py36\n          - rally-tox-py37\n          - rally-dsvm-tox-functional\n          - rally-docker-check\n          - rally-task-basic-with-existing-users:\n              # use_existing_users key did not trigger proper ansible tasks\n              voting: false\n          - rally-task-simple-job\n          - rally-task-barbican:\n              voting: false\n          - rally-task-cinder\n          # NOTE(andreykurilin): this requires more thing to configure before\n          #   launching.\n          #- rally-task-designate\n          - rally-task-heat:\n              voting: false\n          - rally-task-ironic\n          - rally-task-keystone-glance-swift\n          - rally-task-magnum:\n              voting: false\n              files:\n                - rally-jobs/magnum.yaml\n                - rally_openstack/cleanup/resources.py\n                - rally_openstack/contexts/magnum\n                - rally_openstack/scenarios/magnum\n                - rally_openstack/osclients.py\n          - rally-task-manila-no-ss:\n              voting: false\n          - rally-task-manila-ss:\n              voting: false\n          - rally-task-mistral\n          # it did not work for a long time. try to re-configure it\n          #- rally-task-monasca\n          - rally-task-murano:\n              voting: false\n          - rally-task-neutron\n          - rally-task-neutron-trunk:\n              files:\n                - rally-jobs/neutron-trunk.yaml\n                - rally_openstack/scenarios/neutron/trunk.py\n                - rally_openstack/scenarios/neutron/network.py\n          - rally-task-neutron-with-extensions:\n              voting: false\n          - rally-task-nova:\n              voting: false\n          # it did not work for a long time\n          #- rally-task-senlin\n          #- rally-task-octavia:\n          #    voting: false\n          - rally-task-telemetry\n          - rally-task-watcher:\n              # watcher-api did not start last time\n              voting: false\n          - rally-task-zaqar\n          - rally-verify-tempest\n      gate:\n        jobs:\n          - rally-tox-cover\n          - rally-tox-pep8\n          - rally-tox-py27\n          - rally-tox-py35\n          - rally-tox-py36\n          - rally-tox-py37\n          - rally-dsvm-tox-functional\n          - rally-docker-check\n          - rally-task-simple-job\n          - rally-task-cinder\n          #- rally-task-heat\n          - rally-task-ironic\n          - rally-task-keystone-glance-swift\n          - rally-task-mistral\n          - rally-task-neutron\n          - rally-task-telemetry\n          - rally-task-zaqar\n          - rally-verify-tempest\n\n  in \"openstack/rally-openstack/.zuul.d/zuul.yaml@master\", line 12, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f01368aa7fab0d994616ab60547f732fdb87ea2c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-28 13:07:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ab813ab6b43d051f8066a73075b132a96b92a04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:28:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/05d75a354e1941ebb9362474ffa365f3 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902c699664e34718987efeec11ee9e72 : FAILURE in 5m 31s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/06df8b6a0a77469c94488712951f8703 : FAILURE in 4m 32s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ef79ec03b24c42868505cb5ce79196 : SUCCESS in 5m 03s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/650e47a182a54313af4b15618098500d : SUCCESS in 4m 20s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/c92c843262bb4f33b7958260c6407809 : SUCCESS in 5m 08s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/8e7007cb78034091935659efbf0282fc : SUCCESS in 4m 17s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/b0378e335fa0492f9ee8c4fbf462288a : FAILURE in 3m 10s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/dcd51da2d83543e693aa95cfa45af834 : SUCCESS in 7m 32s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fcf830bd750746539ca956c4d45bff04 : SUCCESS in 3m 58s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/302f43184e5d44e6b8c32ccaf40ee7bc : SUCCESS in 3m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/e4327a4b0f6747369860b3d07b6d8c2f : SUCCESS in 5m 03s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/628c5a6468704887a275c41df1d365d4 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":3},{"id":"699207d9ff356f1a131f73d0727c00b0d8094975","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-28 13:44:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7fef95a9e49433af4d2db2abd9fe27839349644f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 14:00:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e42a0701aa7e44ef892a100bd223f782 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a73845d7becf4dc8bbedebdb3454f8bd : SUCCESS in 4m 10s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e1b8342584824e909f72cab0843f7db2 : SUCCESS in 4m 18s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db8ba69f4984c148b029591aeadb206 : FAILURE in 5m 17s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/8628f68273624976b02345ff47995994 : SUCCESS in 4m 38s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/42ae11844ee9422e9159b06d1f2a362e : SUCCESS in 5m 17s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/dade64828889494a8f78ff093369be1c : SUCCESS in 5m 45s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/5ac071b8d8914625943a7bf382df198b : SUCCESS in 5m 53s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/53cf64d09e98439ba621eea1ed9f17f0 : SUCCESS in 7m 52s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/6f3e1f26300047588ccf19b8c3d67657 : SUCCESS in 4m 31s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01b3af67d8bd4d0cba658519b4d0fa6b : SUCCESS in 4m 22s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/8d6687c473e2429591881b1893233ec3 : SUCCESS in 5m 34s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/61f5114583044daebdac565e083fd292 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":4},{"id":"d9df10961326bbcb0fdae77e65af8065cc377170","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-11-28 14:06:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4a27a4a4a2fbfb468d650cca998ac6645956d0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 14:22:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcbbf21fbbd64c37804f83d96836685a : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b846b1fd2dd44dc2b0d86be011b4d551 : SUCCESS in 4m 18s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e7bc222936484136ac870df23e600bdc : SUCCESS in 4m 51s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1faf2a8e67314633a7ed2ae763bc9d66 : SUCCESS in 3m 50s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/7d3a47d9f9d74b8195b781c0114bc1ea : SUCCESS in 5m 40s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7958e92777974ca3b2428f3aa369702c : SUCCESS in 5m 42s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/22312c5bade0448b99c33a35758e5674 : SUCCESS in 4m 25s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9a90de358fdd42b2a85a08ceec1b53ea : SUCCESS in 7m 49s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/a936d877f4a94c479c97bc3a06e6e0b6 : SUCCESS in 8m 25s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/cc71263f58104340ad21bf05227bce05 : SUCCESS in 5m 21s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5639567dacb4dd08d08aa4a2edd560f : SUCCESS in 3m 37s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/194ca2ab594e451daac60bb0f5cc1a1f : SUCCESS in 4m 39s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/64c76e98891b41f8a87f934700204e93 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":5},{"id":"c80f09292ea7d347dec07e03147d9306a469f44c","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 10:14:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1d831d248963c2a6253b63a34dba8da47dc041ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:28:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e157b4bb03640488cbad8238fca32c4 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd599acad4544b997df91610d656466 : SUCCESS in 5m 41s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/4f9ae174ee094beebf1066848dd59138 : SUCCESS in 6m 36s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/1529982432754fbe8b59bbb843a3d613 : SUCCESS in 4m 09s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/14154e17e6664cfda1442cf014e6a190 : SUCCESS in 4m 36s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/545cb249c1a14844822a4a8abf6e3278 : SUCCESS in 5m 53s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/7889882f76a3454ab264c1a4c722bbdc : SUCCESS in 6m 46s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/1adb617e5794445e82b3872d038e7161 : SUCCESS in 6m 50s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/5535f68eba284aeba9533c50da7a2400 : SUCCESS in 8m 44s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/841b47113824459a99b42ef6b7e2d745 : SUCCESS in 4m 20s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c4c1b88fb114a458d762706b1ead74e : SUCCESS in 5m 16s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5ad3d479f57148b6adb234c9447c7d7b : SUCCESS in 5m 28s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/47f16142229143e496ab1d9545c41ba4 : SUCCESS in 7m 45s","accounts_in_message":[],"_revision_number":6},{"id":"2a09cc32f906dd2fe7d03859b91a896f09fa5e37","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 12:52:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c75dd510dd2e66bbe01ffe38a2db3edf8bdeebb8","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 12:53:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c2a965808ffe9f30199aea60c3ad4569d9ba81e3","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 12:53:33.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"426dbb98bc38a0483e24ccdfcac76c46aa4ff376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:16:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bd960887d6344b4f8669eb48cfebce47 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a61143b1d8b4d26a3bca33ce7023ba9 : SUCCESS in 5m 58s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/e607ef77f3b046ed90abccd93db6f853 : SUCCESS in 6m 15s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb9f418247e4cd8967074dfebd12cf6 : SUCCESS in 4m 07s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b82726339d41cea049b01958b9258b : SUCCESS in 4m 03s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/1af421710c6c471390fa12215e77e1b7 : FAILURE in 5m 37s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ad319961034c47e18f04dcda0bff7571 : SUCCESS in 5m 36s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d174028b5d0e420d892d60c5603c2fc4 : SUCCESS in 6m 59s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/c24b707167124f97948cb9cc0b9b6f7a : SUCCESS in 8m 29s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/dfa50298f58d43319b13312b61aed014 : SUCCESS in 5m 57s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48c18e56aa8e4d21965882ca419d4a71 : SUCCESS in 5m 44s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/64232961ce4d4b05beee34c9b32b32b6 : SUCCESS in 5m 07s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/785f3d90ac93410f98fa5d2ef1ef87a1 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":8},{"id":"f238b4873ed025990412831dd9cd577b4e7ab6a2","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 14:08:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2593e53dc5ad99ef129e442b74ccddf42619d6b5","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-02-10 14:19:09.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"f3f9a5b421c33b08c400cf963591aa83e20f626c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:22:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd81d39721214955bd6b353fb617da51 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64bba4cbde3f4878932fda13a34873c4 : SUCCESS in 5m 59s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0c639ee20c8642be9d28522615ba385b : SUCCESS in 5m 28s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/559654f5ad094144823195886f435f5b : SUCCESS in 5m 49s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/275173a258894ce192fd76461ed38b71 : SUCCESS in 7m 24s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/978007872879462da3456d2f94756c7e : SUCCESS in 5m 27s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/ce601a7586d7431db5bce5c699e03817 : SUCCESS in 5m 52s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/9f9fc2b25e7e469ba66d858f217676c8 : SUCCESS in 7m 55s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/a0578af3c3fe41498fd4571a34b6f7b6 : SUCCESS in 9m 19s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/858ad1b2738442eab20ee991683ce162 : SUCCESS in 5m 13s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/894f70c834944fea8f79543ed29db2f4 : SUCCESS in 4m 58s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/a6d100896fb14fcab801f28a5f28cd8e : SUCCESS in 5m 40s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7c134c3599264701b05c35fae8173be5 : SUCCESS in 8m 26s","accounts_in_message":[],"_revision_number":9},{"id":"ad85ff8e16981f391e41a1d26e48043f7a8bc6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:22:59.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"fa5b80be8c3d0a9363d0d4abfccff3e731b6b373","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:39:40.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4c46fc092514266afa97c7b3679035f : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783be4e1a31540efbe0ac6c4ec33723b : SUCCESS in 5m 36s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d6b9416ecc3c4d678c88752f010a83c4 : SUCCESS in 5m 03s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11f0bcef1014edca0f01f6626e2ef8a : SUCCESS in 5m 18s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/0d6fc793fad24954ad4c824d341f8fb6 : SUCCESS in 4m 58s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/7c0112459d1c4d1191e6f295b78d2307 : SUCCESS in 6m 10s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/938c2240465b4795b3e08a5a1a8230b9 : SUCCESS in 7m 53s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/672893281aa44f40886ac76dca61b431 : SUCCESS in 8m 27s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/38c15224b5fd4b1584b710effd401871 : SUCCESS in 11m 33s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/123157c9547f4e28be3ec02734726d18 : SUCCESS in 5m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0363bf4ebc2c4a44ace37f677ec2873e : SUCCESS in 5m 26s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5f553b0764514e3983eb4b5ec7f79f81 : SUCCESS in 8m 20s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/56be3fb576d44023ae4fc0eb0fdbf47c : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":9},{"id":"9143e118034655f58ed29d4673dbcf054f074284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:39:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1e501a6e2175398ffdc506a18e2cb1e1167f8b15","revisions":{"3ccd71aebde8615dc9cc5718cccc1cf625a4bd15":{"kind":"REWORK","_number":1,"created":"2019-11-28 13:00:14.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/1"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"subject":"Stop testing python \u003c3.6","message":"Stop testing python \u003c3.6\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ccd71aebde8615dc9cc5718cccc1cf625a4bd15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/3ccd71aebde8615dc9cc5718cccc1cf625a4bd15"}]},"branch":"refs/heads/master"},"bc2a91b47b36221a3b98fe5e95afcb6b1c51ed8a":{"kind":"REWORK","_number":2,"created":"2019-11-28 13:03:06.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/2"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 13:03:00.000000000","tz":120},"subject":"Stop testing python \u003c3.6","message":"Stop testing python \u003c3.6\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bc2a91b47b36221a3b98fe5e95afcb6b1c51ed8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/bc2a91b47b36221a3b98fe5e95afcb6b1c51ed8a"}]},"branch":"refs/heads/master"},"5f227ee3c370293dbe82c33cdb42cb83a8b4b96e":{"kind":"REWORK","_number":3,"created":"2019-11-28 13:07:39.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/3"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 13:07:22.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5f227ee3c370293dbe82c33cdb42cb83a8b4b96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/5f227ee3c370293dbe82c33cdb42cb83a8b4b96e"}]},"branch":"refs/heads/master"},"de2c3b5868521200aefcbe53fc9e082b959d50d6":{"kind":"REWORK","_number":4,"created":"2019-11-28 13:44:13.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/4"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 13:44:05.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de2c3b5868521200aefcbe53fc9e082b959d50d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/de2c3b5868521200aefcbe53fc9e082b959d50d6"}]},"branch":"refs/heads/master"},"f476e50762d530300c7168c1a03217ca53c74bb3":{"kind":"REWORK","_number":5,"created":"2019-11-28 14:06:10.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/5"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 14:06:03.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f476e50762d530300c7168c1a03217ca53c74bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/f476e50762d530300c7168c1a03217ca53c74bb3"}]},"branch":"refs/heads/master"},"cb5866090eec5028ef5ee5022a39ca2aa90dd46f":{"kind":"REWORK","_number":6,"created":"2020-02-10 10:14:16.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/6"}}},"commit":{"parents":[{"commit":"b5eaf963d0f613644592e45e56706ede6ed3c80f","subject":"Merge \"Fix incorrect __round__ behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b5eaf963d0f613644592e45e56706ede6ed3c80f"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 10:12:15.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cb5866090eec5028ef5ee5022a39ca2aa90dd46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/cb5866090eec5028ef5ee5022a39ca2aa90dd46f"}]},"branch":"refs/heads/master"},"922ca30536be964fa8298929c904aa9a70003763":{"kind":"REWORK","_number":7,"created":"2020-02-10 12:52:18.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/7"}}},"commit":{"parents":[{"commit":"b5eaf963d0f613644592e45e56706ede6ed3c80f","subject":"Merge \"Fix incorrect __round__ behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b5eaf963d0f613644592e45e56706ede6ed3c80f"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 12:51:49.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n+ ignore some warnings during pytest execution to make the output clean\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/922ca30536be964fa8298929c904aa9a70003763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/922ca30536be964fa8298929c904aa9a70003763"}]},"branch":"refs/heads/master"},"c8d14fa1bc561fce08ace6a9074bf5f48cb52984":{"kind":"REWORK","_number":8,"created":"2020-02-10 12:53:26.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/8"}}},"commit":{"parents":[{"commit":"b5eaf963d0f613644592e45e56706ede6ed3c80f","subject":"Merge \"Fix incorrect __round__ behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b5eaf963d0f613644592e45e56706ede6ed3c80f"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 12:53:15.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n+ ignore some warnings during pytest execution to make the output clean\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c8d14fa1bc561fce08ace6a9074bf5f48cb52984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/c8d14fa1bc561fce08ace6a9074bf5f48cb52984"}]},"branch":"refs/heads/master"},"1e501a6e2175398ffdc506a18e2cb1e1167f8b15":{"kind":"REWORK","_number":9,"created":"2020-02-10 14:08:20.000000000","uploader":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"ref":"refs/changes/37/696537/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/37/696537/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/37/696537/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/37/696537/9"}}},"commit":{"parents":[{"commit":"b5eaf963d0f613644592e45e56706ede6ed3c80f","subject":"Merge \"Fix incorrect __round__ behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/b5eaf963d0f613644592e45e56706ede6ed3c80f"}]}],"author":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2019-11-28 12:59:51.000000000","tz":120},"committer":{"name":"Andrey Kurilin","email":"andr.kurilin@gmail.com","date":"2020-02-10 14:08:14.000000000","tz":120},"subject":"Stop supporting python \u003c3.6","message":"Stop supporting python \u003c3.6\n\n+ remove sseveral unused ci files\n+ ignore some warnings during pytest execution to make the output clean\n\nChange-Id: Ic94ab503ea77c8a1582292becad872670034d49c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1e501a6e2175398ffdc506a18e2cb1e1167f8b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/1e501a6e2175398ffdc506a18e2cb1e1167f8b15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
