)]}'
{"id":"openstack%2Frally~696998","triplet_id":"openstack%2Frally~master~I668a4a83b538b49eaae7f3d41de7bbf882827c5b","project":"openstack/rally","branch":"master","hashtags":[],"change_id":"I668a4a83b538b49eaae7f3d41de7bbf882827c5b","subject":"Update the community page.","status":"MERGED","created":"2019-12-03 02:09:36.000000000","updated":"2019-12-04 13:50:15.000000000","submitted":"2019-12-04 13:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696998-1575467415330-73dee3f6","meta_rev_id":"7c57dcbaa71a73e90f2fe25a8e4e7072ca733bd9","_number":696998,"virtual_id_number":696998,"owner":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-04 13:50:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2019-12-04 13:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2019-12-04 13:33:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 13:33:33.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2019-12-04 13:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94608e86d53ef42596fce442ca1ac0cca58db793","author":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"date":"2019-12-03 02:09:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77b16b1ab4afb0839ec9c3dfd79d383e35991757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 02:20:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d468fe78b8e94f5fafa4e9c57bbbf73a : SUCCESS in 3m 49s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/d3b133aee0d84bc2a3d5439a605e66d8 : SUCCESS in 3m 29s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c4af5af27b4a3ab3a880ff64a88c0e : SUCCESS in 3m 31s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/2ccc474ca3d44f44800c99d3365d3756 : SUCCESS in 3m 40s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/980ff25650a64e8e90f0084f1ea21c33 : SUCCESS in 4m 23s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/c7bdf932c2f24ed9a99595e71ec31039 : SUCCESS in 4m 01s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/f3db750df7de4645a61cb65ba3b6d2e5 : SUCCESS in 3m 35s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/4130c1db147d47d88c237a6b9283eebd : SUCCESS in 3m 35s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/610d84a46cb141518990f112254a047f : SUCCESS in 4m 26s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/4b4fd80cb4694a2282dc6775571ec796 : SUCCESS in 5m 37s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/e90a828aa19f43659c10941752b5fd11 : SUCCESS in 5m 05s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/56fe493de2644249b3bb0d58082fde93 : SUCCESS in 7m 12s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f69b1ecac1064481ab4b203638f8aae3 : SUCCESS in 3m 25s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50d0d3c1184a429e88638bb8f4c0b9d8 : SUCCESS in 3m 32s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/5027a16e65364c91865473917ce6565a : SUCCESS in 5m 04s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/7b826a63a6d845da8db911ca117d8835 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"49eef6673061b3fe08b55fc61a257c9a47ad3486","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2019-12-04 13:33:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cf411324ba10e7bcb46b95189916cf28c03f76d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 13:33:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d37da02d4b4bd24e582a9b7901b499644b8b68f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 13:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c57dcbaa71a73e90f2fe25a8e4e7072ca733bd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 13:50:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7364560c6ce149b69596602d2a655632 : SUCCESS in 5m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f86ac8b723ee41fcb28708c962c6f7be : SUCCESS in 4m 32s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f479d6add34d8284e851f5896d171f : SUCCESS in 5m 47s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/aadec2b77474402c89bacc883d1dae72 : SUCCESS in 5m 40s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/5344dd34f3964c848dcea9cfdad610b4 : SUCCESS in 7m 33s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/b6b39c9fbd24442884dbea835eaa70c6 : SUCCESS in 5m 52s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/a51b70e6989546b3a02ed41793965a0b : SUCCESS in 4m 59s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/850500eba7aa4428b82f1894679ebdd3 : SUCCESS in 5m 16s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/2d62c5181e5a4db5ae3a7f1fcff70564 : SUCCESS in 4m 22s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/d5f3c126f2d640bc8f29481efea04c60 : SUCCESS in 6m 42s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/6d333997f3cc4ef5a5ec66d4336b0a29 : SUCCESS in 6m 22s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/a7852e0876b8483d8db17f0aa903aaff : SUCCESS in 7m 52s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/dbd166b202b04fb3a0e43e045a54f7c1 : SUCCESS in 3m 51s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95621f650c274478a4774f3610eb4228 : SUCCESS in 3m 22s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/67072486b60148d6986b15f88b2bf15d : SUCCESS in 4m 51s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/556296c8a764496a9f3d62ecb3be8447 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"028d9b3723fe48da3832ef02b581df8d063eacda","revisions":{"028d9b3723fe48da3832ef02b581df8d063eacda":{"kind":"REWORK","_number":1,"created":"2019-12-03 02:09:36.000000000","uploader":{"_account_id":30717,"name":"wangfaxin","email":"wangfaxin@inspur.com","username":"wangfaxin"},"ref":"refs/changes/98/696998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/98/696998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/696998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/696998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/98/696998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/98/696998/1"}}},"commit":{"parents":[{"commit":"996b060601b5f7e7221aa752da464df08a687d25","subject":"Fix https_key test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/996b060601b5f7e7221aa752da464df08a687d25"}]}],"author":{"name":"wangfaxin","email":"wangfaxin@inspur.com","date":"2019-12-03 02:08:34.000000000","tz":480},"committer":{"name":"wangfaxin","email":"wangfaxin@inspur.com","date":"2019-12-03 02:09:22.000000000","tz":480},"subject":"Update the community page.","message":"Update the community page.\n\nChange-Id: I668a4a83b538b49eaae7f3d41de7bbf882827c5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/028d9b3723fe48da3832ef02b581df8d063eacda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/028d9b3723fe48da3832ef02b581df8d063eacda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
