)]}'
{"id":"openstack%2Frally~702663","triplet_id":"openstack%2Frally~master~I33acb494ddabc6855bd73868ebf8cbeee43bab86","project":"openstack/rally","branch":"master","topic":"add-password-to-user-ctx","hashtags":[],"change_id":"I33acb494ddabc6855bd73868ebf8cbeee43bab86","subject":"Fix incorrect __round__ behaviour","status":"MERGED","created":"2020-01-15 13:13:34.000000000","updated":"2020-01-15 15:07:42.000000000","submitted":"2020-01-15 15:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702663-1579100862393-c94174cc","meta_rev_id":"401b3e00dedc58381317245f8c9eab59c60132f2","_number":702663,"virtual_id_number":702663,"owner":{"_account_id":13009,"name":"Peter Razumovsky","email":"prazumovsky@mirantis.com","username":"prazumovsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 15:07:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":2,"date":"2020-01-15 14:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"all":[{"value":1,"date":"2020-01-15 14:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 14:27:40.000000000","updated_by":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"reviewer":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"state":"REVIEWER"},{"updated":"2020-01-15 15:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16661457e17e3c15a892a40b483985740d9f0357","author":{"_account_id":13009,"name":"Peter Razumovsky","email":"prazumovsky@mirantis.com","username":"prazumovsky"},"date":"2020-01-15 13:13:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7672d315bef5c1b8fb468ba76a4f1103a4c84c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 13:30:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1c53c55c694b5791df28b526a82de7 : SUCCESS in 4m 44s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/f0c4ffbebad94120b0659020409e5604 : SUCCESS in 6m 58s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd2b099a73a41d792b7d56ad29a12d3 : SUCCESS in 5m 41s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/15c5db0dcd604d6bbbfd1535576ae29d : FAILURE in 8m 33s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/682201b5e34547089c80621277804c83 : FAILURE in 4m 43s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/5b937053199c4602a350a06556d15a39 : FAILURE in 5m 27s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/b184dcfee24f4a0ab916e570c2a913d4 : FAILURE in 4m 18s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/a478de85ecc3430194c4e15f6747f997 : FAILURE in 8m 01s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/31d909cb7fea4cb8966ca320f9920150 : FAILURE in 5m 06s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/04aeee3f4d704205b1767a595e2543ee : SUCCESS in 6m 46s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/a31a3ea1f10a4ff8be0bdca908451c9b : SUCCESS in 6m 34s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/55cf6ca33a394b43a55e59bb78cf83aa : SUCCESS in 8m 08s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/5de6367e08c44322bd3cec04fd99d0ed : SUCCESS in 5m 35s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da924c1594d24a01bc3fb95b9907ee31 : SUCCESS in 4m 10s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/4c33c751348c47709f2f1715efdfeaeb : SUCCESS in 5m 08s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/275114dd7347468a8bd0e9d49996fcf3 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d2c351fbdeb9a4870b4952fc33a5a06811c0bbf0","author":{"_account_id":13009,"name":"Peter Razumovsky","email":"prazumovsky@mirantis.com","username":"prazumovsky"},"date":"2020-01-15 13:53:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b7526ca743d2b0100afdecb186a69263c561147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 14:10:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e693892e4742426d8c8a28a4e9512337 : SUCCESS in 5m 13s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0d394b7dd141492a929eaa8712ed7c22 : SUCCESS in 4m 26s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d2d1e43ffbe419b93acbed38ac64cbe : SUCCESS in 5m 09s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/0da4bda33ed04b5b81545feded10f9af : SUCCESS in 5m 15s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/1b7ea6a5e1024e6098b8038e49e57c22 : SUCCESS in 4m 55s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/25cab83a7f7e4412b9fd2f611cf02617 : SUCCESS in 6m 32s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/49d1cc6d120e4a349fed204eb44f3461 : SUCCESS in 4m 29s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/98fa82a48cd144f480241ee43e6b121f : SUCCESS in 5m 30s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/21f485df9cb94d9cbffe4a261e80193b : SUCCESS in 5m 41s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/f5dc883f2e454071965b5e801bc850bd : SUCCESS in 8m 49s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/2b3870654e7740e99f4c46ddfb10328e : SUCCESS in 6m 47s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/0db37698cbd145758f4421727eafd649 : SUCCESS in 9m 17s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/9ac83d5c11cd4084b3be45d8cfa8bfa4 : SUCCESS in 5m 43s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d453ba6c8794ff59825458e13ae12b2 : SUCCESS in 6m 04s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/21e451afc8ef4af1ad577d010fa4934c : SUCCESS in 5m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/4d910d20c1df41c38f06cf39e470336d : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d8d0f3509bf35f8cf71278d3c467c8545ae2529b","author":{"_account_id":9545,"name":"Andriy Kurilin","email":"andr.kurilin@gmail.com","username":"akurilin"},"date":"2020-01-15 14:27:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68184c351bf34277457c027ffe692b160a6a6696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 14:27:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f8c9db526e8993263405480fd65e21271ce2a76c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:07:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d64a3e8ecc17ec2e5f3e7b98ba1931e1ab0aaf68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 15:07:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd3d6643afb470993226dbdf4f45074 : SUCCESS in 4m 03s\n- rally-tox-docs https://zuul.opendev.org/t/openstack/build/0dd3a42e295647de950b76ed4c8d22e9 : SUCCESS in 4m 06s\n- rally-tox-pep8 https://zuul.opendev.org/t/openstack/build/0781b133109c4048a30db874b24e0a94 : SUCCESS in 4m 02s\n- rally-tox-py27 https://zuul.opendev.org/t/openstack/build/ab248df5878247fba145ffaa51f8c686 : SUCCESS in 7m 00s\n- rally-tox-py34 https://zuul.opendev.org/t/openstack/build/8b70f751270647f3bbc5f91afdb97bd7 : SUCCESS in 5m 03s\n- rally-tox-py35 https://zuul.opendev.org/t/openstack/build/d8860be9f0894e4291a2d0c3850558cf : SUCCESS in 4m 53s\n- rally-tox-py36 https://zuul.opendev.org/t/openstack/build/3522b5328f004e3c9eea27a886bbbc81 : SUCCESS in 3m 59s\n- rally-tox-py37 https://zuul.opendev.org/t/openstack/build/019b1784240747738da1b566d3ce87cb : SUCCESS in 7m 55s\n- rally-tox-cover https://zuul.opendev.org/t/openstack/build/650030afce634881b842494fca9aa304 : SUCCESS in 7m 49s\n- rally-tox-functional https://zuul.opendev.org/t/openstack/build/a679af446d2640fab11b76d9de62058a : SUCCESS in 6m 01s\n- rally-tox-functional-py3 https://zuul.opendev.org/t/openstack/build/78502cad6f6a4a8f9072e931479f4e30 : SUCCESS in 6m 09s\n- rally-database-migration https://zuul.opendev.org/t/openstack/build/4c4d213efad24ddea2c6889ad12eba32 : SUCCESS in 7m 40s\n- rally-install-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fc67287f9eb541329ba13f2b2f48dedb : SUCCESS in 4m 12s\n- rally-install-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e3cb61071024be9b48e28097cbe75cb : SUCCESS in 3m 57s\n- rally-install-centos-7 https://zuul.opendev.org/t/openstack/build/d74356ff4b8f4874bfca6b79ef125e74 : SUCCESS in 4m 56s\n- rally-tox-self https://zuul.opendev.org/t/openstack/build/45d2051511c8481c96232bc653ed3d4a : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53241478c822dc48d76f277a6bb91d8ca77e276a","revisions":{"34aab0c0ca9f425ae46e4f29bd11ff5900581455":{"kind":"REWORK","_number":1,"created":"2020-01-15 13:13:34.000000000","uploader":{"_account_id":13009,"name":"Peter Razumovsky","email":"prazumovsky@mirantis.com","username":"prazumovsky"},"ref":"refs/changes/63/702663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/63/702663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/63/702663/1"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"prazumovsky","email":"prazumovsky@mirantis.com","date":"2020-01-15 13:10:07.000000000","tz":240},"committer":{"name":"prazumovsky","email":"prazumovsky@mirantis.com","date":"2020-01-15 13:10:07.000000000","tz":240},"subject":"Fix incorrect __round__ behaviour","message":"Fix incorrect __round__ behaviour\n\nIn some cases with large scenario parameters,\n__round__ raises an error:\n\nTypeError: type NoneType doesn\u0027t define __round__ method\n\nFix __round__ behaviour for StreamingAlgorithm ins \u003d\u003d None.\n\nChange-Id: I33acb494ddabc6855bd73868ebf8cbeee43bab86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/34aab0c0ca9f425ae46e4f29bd11ff5900581455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/34aab0c0ca9f425ae46e4f29bd11ff5900581455"}]},"branch":"refs/heads/master"},"53241478c822dc48d76f277a6bb91d8ca77e276a":{"kind":"REWORK","_number":2,"created":"2020-01-15 13:53:11.000000000","uploader":{"_account_id":13009,"name":"Peter Razumovsky","email":"prazumovsky@mirantis.com","username":"prazumovsky"},"ref":"refs/changes/63/702663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/rally","ref":"refs/changes/63/702663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/rally refs/changes/63/702663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/rally refs/changes/63/702663/2"}}},"commit":{"parents":[{"commit":"21630956e4ff5daa78218c4b4cf4457aefd24d34","subject":"Release 1.5.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/21630956e4ff5daa78218c4b4cf4457aefd24d34"}]}],"author":{"name":"prazumovsky","email":"prazumovsky@mirantis.com","date":"2020-01-15 13:10:07.000000000","tz":240},"committer":{"name":"prazumovsky","email":"prazumovsky@mirantis.com","date":"2020-01-15 13:53:37.000000000","tz":240},"subject":"Fix incorrect __round__ behaviour","message":"Fix incorrect __round__ behaviour\n\nIn some cases with large scenario parameters,\n__round__ raises an error:\n\nTypeError: type NoneType doesn\u0027t define __round__ method\n\nFix __round__ behaviour for StreamingAlgorithm ins \u003d\u003d None.\n\nChange-Id: I33acb494ddabc6855bd73868ebf8cbeee43bab86\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53241478c822dc48d76f277a6bb91d8ca77e276a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/rally/commit/53241478c822dc48d76f277a6bb91d8ca77e276a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
